)]}'
{"id":"openstack%2Ftripleo-validations~663900","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I5e67070e89af5ad3db2c7471e3d55fb704281fc4","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I5e67070e89af5ad3db2c7471e3d55fb704281fc4","subject":"Sync \u0027validations/dhcp-introspection.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:15:04.000000000","updated":"2019-06-09 22:43:39.000000000","submitted":"2019-06-09 22:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663900-1560120219902-26b96e4b","meta_rev_id":"268879b0ab63316a78182ef065181499506dba50","_number":663900,"virtual_id_number":663900,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-09 22:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:18:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:12:04.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:35:55.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:18:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:12:04.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:12:04.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:12:04.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:18:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-09 22:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2872bc5bef46f864e7f7614da5545c532a58ce83","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"730ead5ff9dd0fa33142386484d2f1bb5659f11d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:26:48.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"3f0445a5fcea67db385f610fecea8985009841e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:26:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/663900/1/check/tripleo-ci-centos-7-containers-multinode/5e99b80/ : SUCCESS in 1h 57m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/663900/1/check/tripleo-ci-centos-7-undercloud-containers/f1fbd09/ : SUCCESS in 1h 37m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/663900/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d85e6c8/ : SUCCESS in 2h 13m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/00/663900/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/192c12d/ : SUCCESS in 2h 17m 05s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/663900/1/check/tripleo-ci-centos-7-standalone/065b284/ : SUCCESS in 2h 06m 35s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/663900/1/check/tripleo-ci-fedora-28-standalone/b7e708c/ : SUCCESS in 1h 51m 02s\n- openstack-tox-pep8 http://logs.openstack.org/00/663900/1/check/openstack-tox-pep8/a856375/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/00/663900/1/check/openstack-tox-py27/9d3afbb/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/00/663900/1/check/openstack-tox-py35/e4abb74/ : SUCCESS in 8m 41s\n- openstack-tox-py36 http://logs.openstack.org/00/663900/1/check/openstack-tox-py36/6faa920/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/00/663900/1/check/openstack-tox-docs/24ca300/html/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/663900/1/check/openstack-tox-lower-constraints/4a85ecd/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"eebaed7af4c8c264aea4aed52328214aa65e96f6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:35:55.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"081d855b14ae8660e28e25e00726d2a324abdece","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6461b1b8f7976ac65e1c3cd3654bb1c60c69eb9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:19:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17b0d7e15a19a26db2fa1856ce582342f9cf3f0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11e18246c4e2b1474b7c6acfc7e6da7604de8544","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/663900/1/gate/tripleo-ci-centos-7-containers-multinode/54705ad/ : SUCCESS in 2h 08m 45s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/663900/1/gate/tripleo-ci-centos-7-standalone/26457bb/ : SUCCESS in 1h 46m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/663900/1/gate/tripleo-ci-fedora-28-standalone/8ad929f/ : SUCCESS in 1h 56m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/663900/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/898fa77/ : SUCCESS in 2h 05m 00s\n- openstack-tox-pep8 http://logs.openstack.org/00/663900/1/gate/openstack-tox-pep8/89e7ecb/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/00/663900/1/gate/openstack-tox-py27/9f7a8e5/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/00/663900/1/gate/openstack-tox-py35/75388fe/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/00/663900/1/gate/openstack-tox-py36/4beb775/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/00/663900/1/gate/openstack-tox-docs/228d20b/html/ : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/663900/1/gate/openstack-tox-lower-constraints/a05da2a/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"427f596892206f3bcbe3401cc117ad877936e727","revisions":{"427f596892206f3bcbe3401cc117ad877936e727":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:15:04.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/663900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/663900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/663900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/663900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/663900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/663900/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:14:07.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:14:07.000000000","tz":120},"subject":"Sync \u0027validations/dhcp-introspection.yaml\u0027 with the new role","message":"Sync \u0027validations/dhcp-introspection.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the dhcp-introspection validation.\n\nChange-Id: I5e67070e89af5ad3db2c7471e3d55fb704281fc4\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/427f596892206f3bcbe3401cc117ad877936e727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/427f596892206f3bcbe3401cc117ad877936e727"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
