)]}'
{"id":"openstack%2Ftripleo-validations~663901","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Id1d5bc67697aa939ce35c8a51fd39a573da66eea","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"663072-stable/stein","hashtags":[],"change_id":"Id1d5bc67697aa939ce35c8a51fd39a573da66eea","subject":"Add run-validations.sh script","status":"MERGED","created":"2019-06-07 11:15:27.000000000","updated":"2019-06-12 00:15:19.000000000","submitted":"2019-06-12 00:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663901-1560298519867-216c3fa0","meta_rev_id":"791e81be3335e7cac21a0bd3f2612c2910177ca2","_number":663901,"virtual_id_number":663901,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-11 13:32:59.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-12 00:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"all":[{"value":2,"date":"2019-06-11 13:29:21.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:19:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 23:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:19:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"date":"2019-06-09 08:38:49.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-09 08:38:49.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:19:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 13:29:21.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-11 23:20:46.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-12 00:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5e596a60518fded4e812c5f3a21171484099336","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:15:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ff3da02fdd87bb3884fa92416a47c5e25d0e85d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/check/test-release-openstack/61338c6/ : SUCCESS in 3m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-containers-multinode/ea8c740/ : SUCCESS in 2h 01m 22s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-undercloud-containers/6acce35/ : SUCCESS in 1h 32m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6785fa3/ : SUCCESS in 2h 04m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e110383/ : SUCCESS in 2h 21m 04s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-standalone/e217714/ : SUCCESS in 2h 00m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-fedora-28-standalone/51caaf3/ : SUCCESS in 1h 54m 18s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/check/openstack-tox-pep8/7e883ce/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/check/openstack-tox-py27/1bdc9f0/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/check/openstack-tox-py35/93f073f/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/check/openstack-tox-py36/2ea2e85/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/check/openstack-tox-docs/f5dd1de/html/ : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/check/openstack-tox-lower-constraints/4936ebe/ : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3cd5e5a312a7e9bb6e2109c3f81bcab4ce3ac39b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:19:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3639cf24f6d77565ac972be2478ff174686db7e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ba84a3573b8915dc138ac52ebd415e092dffddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/gate/test-release-openstack/5c7cb0e/ : SUCCESS in 5m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-containers-multinode/b057663/ : SUCCESS in 1h 47m 46s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-standalone/bbc2114/ : SUCCESS in 1h 35m 28s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-fedora-28-standalone/cdc9870/ : FAILURE in 1h 28m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7b060fd/ : SUCCESS in 2h 05m 06s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/gate/openstack-tox-pep8/8cb7dcd/ : SUCCESS in 7m 22s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py27/3b1ca64/ : SUCCESS in 9m 56s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py35/ca50899/ : SUCCESS in 7m 12s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py36/e4904eb/ : SUCCESS in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/gate/openstack-tox-docs/359a925/html/ : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/gate/openstack-tox-lower-constraints/0293ec4/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"381314c759d25665109253e3b6d80159a3f97e4f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-10 09:56:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98b07615dc7c9db298c28ec3458f6eb424fab2e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:39:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/check/test-release-openstack/e2590c2/ : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-containers-multinode/e2664c7/ : SUCCESS in 1h 58m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-undercloud-containers/091656a/ : SUCCESS in 1h 35m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fd58a8e/ : SUCCESS in 2h 16m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9b55453/ : SUCCESS in 2h 39m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-standalone/1e7b35e/ : FAILURE in 1h 54m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-fedora-28-standalone/bfe9f45/ : SUCCESS in 1h 47m 12s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/check/openstack-tox-pep8/d9f27d3/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/check/openstack-tox-py27/54c801f/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/check/openstack-tox-py35/8130277/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/check/openstack-tox-py36/2c617d7/ : SUCCESS in 6m 31s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/check/openstack-tox-docs/3d22cff/html/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/check/openstack-tox-lower-constraints/49abe2c/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"38dc03990c37c4008dfb0f23532c437f7a9eddbd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:35:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"783de0ef5dae3ae94d2cc3383d43105dcbccdd4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:26:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/check/test-release-openstack/261c068/ : SUCCESS in 3m 32s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-containers-multinode/42e2560/ : SUCCESS in 2h 11m 56s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-undercloud-containers/a525514/ : SUCCESS in 1h 28m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/76f548b/ : SUCCESS in 2h 23m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a673fe8/ : SUCCESS in 2h 44m 07s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-standalone/b0d468f/ : SUCCESS in 1h 37m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-fedora-28-standalone/30d00ec/ : SUCCESS in 1h 48m 33s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/check/openstack-tox-pep8/0f85de2/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/check/openstack-tox-py27/e78d64a/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/check/openstack-tox-py35/d71fc2b/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/check/openstack-tox-py36/ff858b8/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/check/openstack-tox-docs/948d96b/html/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/check/openstack-tox-lower-constraints/805cacd/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7b4974003979e91c8b89ef6c21dc8450e586d00e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36bec4805d5e78668700b09179bbc3e3b2c45477","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:11:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/gate/test-release-openstack/39d44a4/ : SUCCESS in 4m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-containers-multinode/e710ab0/ : FAILURE in 17m 59s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-standalone/a9960ea/ : SUCCESS in 2h 05m 49s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-fedora-28-standalone/0add436/ : SUCCESS in 1h 56m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5018cbd/ : SUCCESS in 2h 25m 15s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/gate/openstack-tox-pep8/31c2072/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py27/a7fb38d/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py35/774d8dd/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py36/ffff1da/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/gate/openstack-tox-docs/55aff20/html/ : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/gate/openstack-tox-lower-constraints/48b3558/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"619cc90f9605967dd3911e0c7ce7bab370cac0c1","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-11 12:52:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"335d9f779f766ab1a96e6a46f177129d462f9bda","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2019-06-11 13:29:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e46b1889bb26a27befba97de0c2a3028ed06fbb6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-11 13:32:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4063f4eab34d43015a21e933d7dfd04a625cc3f7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-11 13:57:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5376948f4cf04a7ea2f64f850435b97b23c3d119","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/check/test-release-openstack/29adfd7/ : SUCCESS in 3m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-containers-multinode/bca3504/ : SUCCESS in 2h 02m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-undercloud-containers/411776f/ : SUCCESS in 1h 36m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bb1458d/ : SUCCESS in 2h 23m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/171aa20/ : SUCCESS in 2h 37m 52s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-centos-7-standalone/9581140/ : SUCCESS in 1h 39m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/check/tripleo-ci-fedora-28-standalone/265a556/ : SUCCESS in 1h 53m 37s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/check/openstack-tox-pep8/5b35016/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/check/openstack-tox-py27/f526b85/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/check/openstack-tox-py35/73e3f23/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/check/openstack-tox-py36/1bfd0ea/ : SUCCESS in 5m 40s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/check/openstack-tox-docs/b0a7aea/html/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/check/openstack-tox-lower-constraints/1aea6c6/ : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0f7ff2a5de3aa6b9fe4dae01cd85d4b768287c24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:46:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bae6b8c4ef3918d37426a5af3998e032204af45","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-11 23:20:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2531230e3ddf921f5b56ad59f0a9a818940420c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:15:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1e9b80b2e2464564b7f56a1685230e3ad58951d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:15:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/01/663901/1/gate/test-release-openstack/4ca97b8/ : SUCCESS in 3m 23s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-containers-multinode/2533763/ : SUCCESS in 2h 10m 16s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-standalone/e9667bb/ : SUCCESS in 1h 57m 35s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/01/663901/1/gate/tripleo-ci-fedora-28-standalone/afc66af/ : SUCCESS in 1h 40m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/01/663901/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4439ad9/ : SUCCESS in 2h 07m 23s\n- openstack-tox-pep8 http://logs.openstack.org/01/663901/1/gate/openstack-tox-pep8/786dead/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py27/f973fda/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py35/7313784/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/01/663901/1/gate/openstack-tox-py36/4d5a72d/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/01/663901/1/gate/openstack-tox-docs/deedd07/html/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/663901/1/gate/openstack-tox-lower-constraints/c8599b2/ : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40f33af33d26c0593110b981fba310bf5d6f8477","revisions":{"40f33af33d26c0593110b981fba310bf5d6f8477":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:15:27.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/01/663901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/01/663901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/663901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/663901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/663901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/01/663901/1"}}},"commit":{"parents":[{"commit":"9d3d67647a8e93c495bcdcbdbaaa77fa64db069b","subject":"New validation: ensure image-serve service is working as expected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9d3d67647a8e93c495bcdcbdbaaa77fa64db069b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-04 15:47:48.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:15:27.000000000","tz":0},"subject":"Add run-validations.sh script","message":"Add run-validations.sh script\n\nThis patch adds a simple script to run a validation through the\nansible-playbook command-line and only for testing-purpose.\n\nChange-Id: Id1d5bc67697aa939ce35c8a51fd39a573da66eea\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 7c729cc83e99de09995638e2728901469887c70b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/40f33af33d26c0593110b981fba310bf5d6f8477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/40f33af33d26c0593110b981fba310bf5d6f8477"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
