)]}'
{"id":"openstack%2Ftripleo-validations~663905","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I1b8e483ed4cb1ea116216a08baae267b2fd8b2e1","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I1b8e483ed4cb1ea116216a08baae267b2fd8b2e1","subject":"Sync \u0027validations/ironic-boot-configuration.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:25:05.000000000","updated":"2019-06-09 23:58:44.000000000","submitted":"2019-06-09 23:58:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663905-1560124724557-0eece10d","meta_rev_id":"059e2717f47085d1f8cdbabebb15613da84149c6","_number":663905,"virtual_id_number":663905,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-09 23:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:19:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:12:00.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:35:20.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:19:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:11:59.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:59.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 12:12:00.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-09 16:19:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-09 23:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50983abb4b0a3b974ceebddab73055117974cb1e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18e68f9399bc5e873d06349ec017ae9824606f5c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:27:43.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"8227d5b99a68855982fc30f69f50853ce2552957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:50:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/663905/1/check/tripleo-ci-centos-7-containers-multinode/f85e643/ : SUCCESS in 2h 22m 40s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/663905/1/check/tripleo-ci-centos-7-undercloud-containers/40021f6/ : SUCCESS in 1h 44m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/663905/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f94aaca/ : SUCCESS in 2h 28m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/663905/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ea14ddb/ : SUCCESS in 2h 40m 57s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/663905/1/check/tripleo-ci-centos-7-standalone/8954ab9/ : SUCCESS in 1h 37m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/663905/1/check/tripleo-ci-fedora-28-standalone/c44d9d0/ : SUCCESS in 1h 54m 31s\n- openstack-tox-pep8 http://logs.openstack.org/05/663905/1/check/openstack-tox-pep8/931b0f9/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/05/663905/1/check/openstack-tox-py27/bdd2461/ : SUCCESS in 6m 41s\n- openstack-tox-py35 http://logs.openstack.org/05/663905/1/check/openstack-tox-py35/74b8dbd/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/05/663905/1/check/openstack-tox-py36/276f9ff/ : SUCCESS in 6m 33s\n- openstack-tox-docs http://logs.openstack.org/05/663905/1/check/openstack-tox-docs/f6b5dd5/html/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/663905/1/check/openstack-tox-lower-constraints/a26e458/ : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b94fa6d85ef27a841554cbc742b50bde66a89745","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:35:20.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"27e8abb244043cfd6536c476ff5c36d8351c3b6e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:19:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9187fbff2767e67117691924a307c513048997f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e906dcd345d33d76dd5cbbc05882985cb176262d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 23:58:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/663905/1/gate/tripleo-ci-centos-7-containers-multinode/5eefe9e/ : SUCCESS in 2h 21m 01s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/663905/1/gate/tripleo-ci-centos-7-standalone/c7dd12e/ : SUCCESS in 1h 59m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/663905/1/gate/tripleo-ci-fedora-28-standalone/598162f/ : SUCCESS in 2h 00m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/663905/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/edac0a5/ : SUCCESS in 2h 23m 48s\n- openstack-tox-pep8 http://logs.openstack.org/05/663905/1/gate/openstack-tox-pep8/6bbf8c8/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/05/663905/1/gate/openstack-tox-py27/4546b80/ : SUCCESS in 7m 17s\n- openstack-tox-py35 http://logs.openstack.org/05/663905/1/gate/openstack-tox-py35/ab1cd66/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/05/663905/1/gate/openstack-tox-py36/9a62759/ : SUCCESS in 7m 33s\n- openstack-tox-docs http://logs.openstack.org/05/663905/1/gate/openstack-tox-docs/3bcfdd7/html/ : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/663905/1/gate/openstack-tox-lower-constraints/ca277b4/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ced0cd79a958c3e93050785502692218c737c881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 23:58:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea91001925b8c2deb2da858e9dff50a2b737df76","revisions":{"ea91001925b8c2deb2da858e9dff50a2b737df76":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:25:05.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/05/663905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/05/663905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/663905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/663905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/663905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/05/663905/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:24:11.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:24:11.000000000","tz":120},"subject":"Sync \u0027validations/ironic-boot-configuration.yaml\u0027 with the new role","message":"Sync \u0027validations/ironic-boot-configuration.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the ironic-boot-configuration validation.\n\nChange-Id: I1b8e483ed4cb1ea116216a08baae267b2fd8b2e1\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ea91001925b8c2deb2da858e9dff50a2b737df76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ea91001925b8c2deb2da858e9dff50a2b737df76"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
