)]}'
{"id":"openstack%2Ftripleo-validations~663907","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ife345309d3bbf29575d5f14a4e1935ef3462701d","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"Ife345309d3bbf29575d5f14a4e1935ef3462701d","subject":"Sync \u0027validations/controller-token.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:33:42.000000000","updated":"2019-06-11 10:04:32.000000000","submitted":"2019-06-11 10:04:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663907-1560247472763-d9eb55b5","meta_rev_id":"955db214a9ef2545f408909d69223026c206031d","_number":663907,"virtual_id_number":663907,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 10:04:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:20:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:11:55.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-10 09:56:05.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:20:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:11:55.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:55.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:11:55.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:20:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 10:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0a716c764089f8dd3fd047b3ad3c315e3a5d559","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e282058e22c5a69f7140fb7b5f382954fb3ef2c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:28:13.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"01ff05860e1888ab30e627dead6d7d5a91e5265a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-containers-multinode/0111fd7/ : SUCCESS in 2h 11m 06s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-undercloud-containers/a9c17c4/ : SUCCESS in 1h 28m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f73047f/ : SUCCESS in 2h 14m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d715baf/ : SUCCESS in 2h 28m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-standalone/cf799c1/ : SUCCESS in 1h 49m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-fedora-28-standalone/c01fd71/ : SUCCESS in 1h 46m 17s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/check/openstack-tox-pep8/afd0660/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/check/openstack-tox-py27/0cfc7d7/ : SUCCESS in 6m 15s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/check/openstack-tox-py35/7ea8f11/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/check/openstack-tox-py36/7120764/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/check/openstack-tox-docs/0c37b20/html/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/check/openstack-tox-lower-constraints/d571d6c/ : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b70935d634129c08bf3aa06e3d70e7c25f57a76e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:35:26.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"2a65c4e963018b6188bf037ef74f4e3098fc5919","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:20:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff5ecf1d3150f332f32c59562da06b3540cc7648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99987d2cc1bf805ad776162989073ac247ef9f80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 00:18:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-containers-multinode/c8114b4/ : FAILURE in 1h 47m 12s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-standalone/f0f7fbc/ : SUCCESS in 1h 54m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-fedora-28-standalone/9fec116/ : SUCCESS in 1h 31m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4fdb558/ : SUCCESS in 2h 03m 18s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/gate/openstack-tox-pep8/d8d9155/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py27/46fe21b/ : SUCCESS in 7m 37s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py35/1103ee9/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py36/313329a/ : SUCCESS in 6m 27s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/gate/openstack-tox-docs/ed938f1/html/ : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/gate/openstack-tox-lower-constraints/5a399cf/ : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3817a5b2dd87997e2952cd9f3d5ff91f32d9ff35","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-10 09:56:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49e377386b2129e22602e1aaba18b85d1b2a48a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:41:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-containers-multinode/ae787a6/ : SUCCESS in 1h 50m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-undercloud-containers/7ac1521/ : SUCCESS in 1h 27m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/82d2bb3/ : SUCCESS in 2h 11m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2d8c92d/ : SUCCESS in 2h 38m 57s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-standalone/dc6ebdf/ : SUCCESS in 1h 56m 55s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-fedora-28-standalone/7fdcf59/ : SUCCESS in 2h 06m 24s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/check/openstack-tox-pep8/90f72ca/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/check/openstack-tox-py27/bef69a7/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/check/openstack-tox-py35/ca75f0e/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/check/openstack-tox-py36/e2837e2/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/check/openstack-tox-docs/e7a20ca/html/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/check/openstack-tox-lower-constraints/bec5cec/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c9242e04a53ce3d469f02a86a456893bcf999ec7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:41:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad2cf2023cbf9de7bc48c3d51f93e4d34ab6ac65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:27:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-containers-multinode/33f7db8/ : SUCCESS in 2h 26m 05s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-standalone/96be090/ : FAILURE in 1h 57m 41s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-fedora-28-standalone/d8fb7b2/ : SUCCESS in 2h 13m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3bfbf92/ : SUCCESS in 2h 27m 14s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/gate/openstack-tox-pep8/ed80a65/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py27/437b0c4/ : SUCCESS in 6m 49s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py35/4ea335a/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py36/af525b4/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/gate/openstack-tox-docs/a7e8e42/html/ : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/gate/openstack-tox-lower-constraints/dd4d25c/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cec299bde2e3356ee58f68d65f1f52b2db03f9ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 20:06:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8106f54bd52303fd2fca14e0b6376b585dd97d57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 02:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-containers-multinode/3f5b1b3/ : SUCCESS in 2h 04m 46s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-undercloud-containers/7be1901/ : SUCCESS in 1h 35m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/352bb2f/ : SUCCESS in 2h 15m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3e42dc4/ : SUCCESS in 2h 22m 05s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-centos-7-standalone/1651e38/ : SUCCESS in 1h 51m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/check/tripleo-ci-fedora-28-standalone/be321f8/ : SUCCESS in 1h 51m 25s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/check/openstack-tox-pep8/59d0b39/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/check/openstack-tox-py27/b19cabb/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/check/openstack-tox-py35/a916ce4/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/check/openstack-tox-py36/0cadaba/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/check/openstack-tox-docs/2617f79/html/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/check/openstack-tox-lower-constraints/1e7880e/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3f855255f4b2eba602ccee40dfaf0b9ec7d8f35d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4398afff5ee0ca30c27c827a63012499568275c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-containers-multinode/65ed223/ : SUCCESS in 2h 02m 09s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-standalone/16c1e62/ : SUCCESS in 2h 15m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/663907/1/gate/tripleo-ci-fedora-28-standalone/cc94572/ : SUCCESS in 2h 08m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/663907/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/586b22a/ : SUCCESS in 2h 27m 19s\n- openstack-tox-pep8 http://logs.openstack.org/07/663907/1/gate/openstack-tox-pep8/78ea38a/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py27/40119da/ : SUCCESS in 5m 25s\n- openstack-tox-py35 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py35/1a15528/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/07/663907/1/gate/openstack-tox-py36/4c388d4/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/07/663907/1/gate/openstack-tox-docs/5150a24/html/ : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/663907/1/gate/openstack-tox-lower-constraints/0ec31f4/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ca279929bd3537b1e9f011d9df41acff3e5bfd45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"983e33f6f391610f5f5cbbf4b1bd2e12b26a2537","revisions":{"983e33f6f391610f5f5cbbf4b1bd2e12b26a2537":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:33:42.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/663907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/663907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/663907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/663907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/663907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/663907/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:32:55.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:32:55.000000000","tz":120},"subject":"Sync \u0027validations/controller-token.yaml\u0027 with the new role","message":"Sync \u0027validations/controller-token.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the controller-token validation.\n\nChange-Id: Ife345309d3bbf29575d5f14a4e1935ef3462701d\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/983e33f6f391610f5f5cbbf4b1bd2e12b26a2537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/983e33f6f391610f5f5cbbf4b1bd2e12b26a2537"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
