)]}'
{"id":"openstack%2Ftripleo-validations~663912","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I614a739d7b9976af47f0d175d44ad525705d4453","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I614a739d7b9976af47f0d175d44ad525705d4453","subject":"Sync \u0027validations/ceilometerdb-size.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:48:52.000000000","updated":"2019-06-10 02:17:18.000000000","submitted":"2019-06-10 02:17:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663912-1560133038397-f0d02f45","meta_rev_id":"b11b4203ae9f3f5ba32e6e2b1de0c39dff26c2a2","_number":663912,"virtual_id_number":663912,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 02:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:20:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:11:49.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:34:34.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:20:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:11:49.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:49.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:11:49.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:20:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 02:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c7678909f3fd6c3d444f17066ba75157791c3a5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:48:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20cb51b1bc672139662aba4c362b4da54f73f032","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:29:27.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"14b2cf6bc2d2e142fb0cf234e72923e6cd8220fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:02:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/12/663912/1/check/tripleo-ci-centos-7-containers-multinode/33d0d6a/ : SUCCESS in 2h 01m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/12/663912/1/check/tripleo-ci-centos-7-undercloud-containers/7ed2519/ : SUCCESS in 1h 27m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/12/663912/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d13ed43/ : SUCCESS in 2h 07m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/12/663912/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0e5f241/ : SUCCESS in 2h 37m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/12/663912/1/check/tripleo-ci-centos-7-standalone/976ad95/ : SUCCESS in 1h 50m 14s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/12/663912/1/check/tripleo-ci-fedora-28-standalone/a182f45/ : SUCCESS in 1h 50m 52s\n- openstack-tox-pep8 http://logs.openstack.org/12/663912/1/check/openstack-tox-pep8/d725ec4/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/12/663912/1/check/openstack-tox-py27/0a65405/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/12/663912/1/check/openstack-tox-py35/0de5b5b/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/12/663912/1/check/openstack-tox-py36/aa17f38/ : SUCCESS in 6m 20s\n- openstack-tox-docs http://logs.openstack.org/12/663912/1/check/openstack-tox-docs/65cadf7/html/ : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/663912/1/check/openstack-tox-lower-constraints/f361fe8/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"589cd96b6301321210fae896c1190c072e6be1f2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:34:34.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"8c5807a146785592b2a51f7b7eeaf618f2b4a1ed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:20:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1047b283235ae7931ebbb09aaa32e6b5fcd424d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9acc25cc48209d4ad6a0353a7cd24141970afca8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:17:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b185797123854c2ffc51946bec9717c95b3cc238","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:17:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/12/663912/1/gate/tripleo-ci-centos-7-containers-multinode/b76b6cb/ : SUCCESS in 2h 00m 33s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/12/663912/1/gate/tripleo-ci-centos-7-standalone/466f6b2/ : SUCCESS in 1h 44m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/12/663912/1/gate/tripleo-ci-fedora-28-standalone/b91510e/ : SUCCESS in 1h 42m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/12/663912/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d5dbfe3/ : SUCCESS in 2h 16m 46s\n- openstack-tox-pep8 http://logs.openstack.org/12/663912/1/gate/openstack-tox-pep8/04dbbbf/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/12/663912/1/gate/openstack-tox-py27/fbeae9c/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/12/663912/1/gate/openstack-tox-py35/13423e2/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/12/663912/1/gate/openstack-tox-py36/d55bdac/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/12/663912/1/gate/openstack-tox-docs/e198962/html/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/663912/1/gate/openstack-tox-lower-constraints/fa78561/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"472d4a89f11fff93f6bb202eb25d3531fdd65b57","revisions":{"472d4a89f11fff93f6bb202eb25d3531fdd65b57":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:48:52.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/12/663912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/12/663912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/663912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/663912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/663912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/12/663912/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:47:51.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:47:51.000000000","tz":120},"subject":"Sync \u0027validations/ceilometerdb-size.yaml\u0027 with the new role","message":"Sync \u0027validations/ceilometerdb-size.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the ceilometerdb-size validations.\n\nChange-Id: I614a739d7b9976af47f0d175d44ad525705d4453\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/472d4a89f11fff93f6bb202eb25d3531fdd65b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/472d4a89f11fff93f6bb202eb25d3531fdd65b57"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
