)]}'
{"id":"openstack%2Ftripleo-validations~663913","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I7bb40a1f0b42cbbf5b38a3939db9140b602e5c52","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I7bb40a1f0b42cbbf5b38a3939db9140b602e5c52","subject":"Sync \u0027validations/check-network-gateway.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:52:20.000000000","updated":"2019-06-12 00:00:22.000000000","submitted":"2019-06-12 00:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663913-1560297622073-d90c8b36","meta_rev_id":"d62b234e1281805c3385ebc5c50f9e2ac5b1727c","_number":663913,"virtual_id_number":663913,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-12 00:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2019-06-11 12:52:29.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-11 12:30:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 12:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-11 12:52:29.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:46.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 12:11:47.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-09 23:13:39.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-10 13:15:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-12 00:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b614d725eec3fcb3876e3d1359a367dae6b0b80b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9544850d880a6a81a2ec3f312d2a69d43c04c6c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:29:42.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"d1b408af2e8f1ccf4ae54076e9d5c3607ea55cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:12:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-containers-multinode/dab7d11/ : SUCCESS in 2h 01m 36s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-undercloud-containers/a553ecb/ : SUCCESS in 1h 30m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7e8d3fe/ : SUCCESS in 2h 17m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6929540/ : SUCCESS in 2h 22m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-standalone/1925167/ : FAILURE in 1h 44m 43s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/663913/1/check/tripleo-ci-fedora-28-standalone/2efba73/ : SUCCESS in 1h 48m 10s\n- openstack-tox-pep8 http://logs.openstack.org/13/663913/1/check/openstack-tox-pep8/a9e0f45/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/13/663913/1/check/openstack-tox-py27/8450d73/ : SUCCESS in 6m 33s\n- openstack-tox-py35 http://logs.openstack.org/13/663913/1/check/openstack-tox-py35/eb3976e/ : SUCCESS in 7m 54s\n- openstack-tox-py36 http://logs.openstack.org/13/663913/1/check/openstack-tox-py36/6c06751/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/13/663913/1/check/openstack-tox-docs/1f3cead/html/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/663913/1/check/openstack-tox-lower-constraints/4b60d0b/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"60fa12e894ba2cb7a1b0b76433488f91d7bff4c6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:34:26.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"e4bcbe7c8831c015842899ef2132f113b867b455","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:20:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4348263e561b7f34ec80bd4172cb04ea0c714543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 18:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-containers-multinode/d5e788b/ : SUCCESS in 2h 05m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-undercloud-containers/d3f6026/ : SUCCESS in 1h 36m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6c56614/ : SUCCESS in 2h 20m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/91dd776/ : SUCCESS in 2h 25m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/663913/1/check/tripleo-ci-centos-7-standalone/4d42729/ : SUCCESS in 1h 56m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/663913/1/check/tripleo-ci-fedora-28-standalone/cb15fa4/ : SUCCESS in 1h 48m 20s\n- openstack-tox-pep8 http://logs.openstack.org/13/663913/1/check/openstack-tox-pep8/48295b8/ : SUCCESS in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/13/663913/1/check/openstack-tox-py27/94248c1/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/13/663913/1/check/openstack-tox-py35/1370c8e/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/13/663913/1/check/openstack-tox-py36/6fae4b1/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/13/663913/1/check/openstack-tox-docs/fa13158/html/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/663913/1/check/openstack-tox-lower-constraints/da09eee/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a7f57a18570196d10a680e4798f1641c5bcfc86d","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-09 23:13:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3173badf6cf3a618e1b88fb218d33d2ac10db212","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:15:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12aca08dda03986acb5ce57f8c6a7ac6df98c4c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:15:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fae54dc069178cbee42c1599a18a833bd5d6eb75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:19:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/663913/1/gate/tripleo-ci-centos-7-containers-multinode/c30f312/ : SUCCESS in 2h 00m 42s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/663913/1/gate/tripleo-ci-centos-7-standalone/b27731e/ : SUCCESS in 1h 55m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/663913/1/gate/tripleo-ci-fedora-28-standalone/c8fd980/ : FAILURE in 2h 04m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/663913/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4254e9d/ : SUCCESS in 2h 21m 12s\n- openstack-tox-pep8 http://logs.openstack.org/13/663913/1/gate/openstack-tox-pep8/4c0f3ef/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/13/663913/1/gate/openstack-tox-py27/a6807d9/ : SUCCESS in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/13/663913/1/gate/openstack-tox-py35/5c5b184/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/13/663913/1/gate/openstack-tox-py36/9e979a4/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/13/663913/1/gate/openstack-tox-docs/3ed5112/html/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/663913/1/gate/openstack-tox-lower-constraints/fef38da/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6205319552c36e848bd35135506907b6a4d92888","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-11 09:12:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e93dce140747ead0009d5343b5c87713c014e30b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-11 12:30:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d27875efa05ce8415768901e8ae13bb84b6ee285","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-11 12:52:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8291fd9768e94a40e4c83a3cc01ad9238a7c5a20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:33:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/663913/2/check/tripleo-ci-centos-7-containers-multinode/05f1807/ : SUCCESS in 2h 09m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/13/663913/2/check/tripleo-ci-centos-7-undercloud-containers/de6923a/ : SUCCESS in 1h 37m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/663913/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b6e791a/ : SUCCESS in 2h 23m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/13/663913/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/dfe6554/ : SUCCESS in 2h 24m 36s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/663913/2/check/tripleo-ci-centos-7-standalone/b54a554/ : SUCCESS in 2h 16m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/663913/2/check/tripleo-ci-fedora-28-standalone/905429d/ : SUCCESS in 1h 48m 39s\n- openstack-tox-pep8 http://logs.openstack.org/13/663913/2/check/openstack-tox-pep8/b9d4668/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/13/663913/2/check/openstack-tox-py27/630c4b1/ : SUCCESS in 5m 50s\n- openstack-tox-py35 http://logs.openstack.org/13/663913/2/check/openstack-tox-py35/8a7571a/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/13/663913/2/check/openstack-tox-py36/00ba7d4/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/13/663913/2/check/openstack-tox-docs/4f2329c/html/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/663913/2/check/openstack-tox-lower-constraints/ba02d59/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1c3821dfb4e75bf93de54ed8cd3f048d248e8a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:33:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfdf043204a5e583fbd0c49b4ad77ad423bca273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:00:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/663913/2/gate/tripleo-ci-centos-7-containers-multinode/de6b53c/ : SUCCESS in 2h 02m 27s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/663913/2/gate/tripleo-ci-centos-7-standalone/4bb5a72/ : SUCCESS in 1h 42m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/663913/2/gate/tripleo-ci-fedora-28-standalone/69c5012/ : SUCCESS in 1h 49m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/663913/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/dc8fe3d/ : SUCCESS in 2h 20m 40s\n- openstack-tox-pep8 http://logs.openstack.org/13/663913/2/gate/openstack-tox-pep8/30311f2/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/13/663913/2/gate/openstack-tox-py27/791916f/ : SUCCESS in 6m 32s\n- openstack-tox-py35 http://logs.openstack.org/13/663913/2/gate/openstack-tox-py35/bd2c09d/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/13/663913/2/gate/openstack-tox-py36/8e81af5/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/13/663913/2/gate/openstack-tox-docs/86c30d1/html/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/663913/2/gate/openstack-tox-lower-constraints/0474186/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"35d0b3f3ff8642440f35202252337983b1bff81a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:00:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5e22f882ef1a2fa20b5afbef03e4e3aec5a1f0d","revisions":{"5d13f40d86ba553736be29bd2a475df526f8e9ea":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:52:20.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/13/663913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/13/663913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:51:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:51:20.000000000","tz":120},"subject":"Sync \u0027validations/check-network-gateway.yaml\u0027 with the new role","message":"Sync \u0027validations/check-network-gateway.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the check-network-gateway validation.\n\nChange-Id: I7bb40a1f0b42cbbf5b38a3939db9140b602e5c52\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5d13f40d86ba553736be29bd2a475df526f8e9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5d13f40d86ba553736be29bd2a475df526f8e9ea"}]},"branch":"refs/heads/stable/stein"},"d5e22f882ef1a2fa20b5afbef03e4e3aec5a1f0d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-11 12:30:25.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/13/663913/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/13/663913/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/13/663913/2"}}},"commit":{"parents":[{"commit":"a3348f8545c1024b46a3e1c635d275f60c2a6bfb","subject":"Merge \"Fix typo in the command to get the file_descriptors total_limit\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a3348f8545c1024b46a3e1c635d275f60c2a6bfb"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:51:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-11 12:30:25.000000000","tz":0},"subject":"Sync \u0027validations/check-network-gateway.yaml\u0027 with the new role","message":"Sync \u0027validations/check-network-gateway.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the check-network-gateway validation.\n\nChange-Id: I7bb40a1f0b42cbbf5b38a3939db9140b602e5c52\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d5e22f882ef1a2fa20b5afbef03e4e3aec5a1f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d5e22f882ef1a2fa20b5afbef03e4e3aec5a1f0d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
