)]}'
{"id":"openstack%2Ftripleo-validations~663918","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I6555d97051b4c0a1361e69e24ef849bf98828e0b","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I6555d97051b4c0a1361e69e24ef849bf98828e0b","subject":"Sync \u0027validations/callback_plugins/validation_output.py\u0027","status":"MERGED","created":"2019-06-07 12:10:39.000000000","updated":"2019-06-11 10:04:29.000000000","submitted":"2019-06-11 10:04:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663918-1560247469571-8e310be5","meta_rev_id":"9e95f51013ae6b69fca9e7ff95a7717a5ea506d5","_number":663918,"virtual_id_number":663918,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 10:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:20:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:11:44.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-10 09:56:31.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:20:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:11:44.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:11:44.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:11:44.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:20:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 10:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"450bd8b15885c9f250a9ba271e7cf2e2fa3ef0bd","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 12:10:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c72c5a137bde423235588d508274faa1a4dabcea","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:29:58.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"8911c9b8cfe8558bde28f932e26d3f2545a7ac27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:35:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-containers-multinode/ac2362e/ : SUCCESS in 2h 10m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-undercloud-containers/fb6a3da/ : SUCCESS in 1h 32m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d454fee/ : SUCCESS in 2h 09m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9a519ec/ : SUCCESS in 2h 36m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-standalone/8a182c5/ : SUCCESS in 1h 49m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-fedora-28-standalone/268d086/ : SUCCESS in 1h 56m 11s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/check/openstack-tox-pep8/4d3974c/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/check/openstack-tox-py27/5c75858/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/check/openstack-tox-py35/5252d90/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/check/openstack-tox-py36/661ca0f/ : SUCCESS in 7m 11s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/check/openstack-tox-docs/1a7b5db/html/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/check/openstack-tox-lower-constraints/464d936/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2d2e2d97f57db36340e56d9f600d1df7a3688625","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:33:59.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ebb5aeaba7cdcf63563a1a7b4063c3e292c796fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:20:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e70686c083b6240e6429636b19f77a0dde70dbb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9275dbe9b56a2967503ace3bdd6907ad3a0b717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:17:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-containers-multinode/5c7f267/ : SUCCESS in 1h 57m 45s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-standalone/74897bf/ : SUCCESS in 1h 52m 28s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-fedora-28-standalone/91662eb/ : FAILURE in 1h 54m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f898b75/ : SUCCESS in 2h 02m 38s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/gate/openstack-tox-pep8/f590b39/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py27/3f5bae4/ : SUCCESS in 9m 03s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py35/85da49d/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py36/9e236c4/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/gate/openstack-tox-docs/fc80ac3/html/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/gate/openstack-tox-lower-constraints/7352d46/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a379b3ace5136d04aa0a8cd75e5b1b6df7c8db4c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-10 09:56:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1865b62c8d32ce9b4faa4b66e4aa59c8654d8e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:27:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-containers-multinode/20e9d90/ : SUCCESS in 1h 53m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-undercloud-containers/286597f/ : SUCCESS in 1h 31m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2261e0d/ : SUCCESS in 2h 19m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/869f4b8/ : SUCCESS in 2h 17m 31s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-standalone/a705f99/ : SUCCESS in 1h 57m 56s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-fedora-28-standalone/37aac8e/ : SUCCESS in 1h 47m 54s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/check/openstack-tox-pep8/55e3bb2/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/check/openstack-tox-py27/b2aa73e/ : SUCCESS in 5m 22s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/check/openstack-tox-py35/e589796/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/check/openstack-tox-py36/1266eb6/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/check/openstack-tox-docs/be8e9b4/html/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/check/openstack-tox-lower-constraints/2d8acf6/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"871012a17a2ce79fa9035c99e298ad463113308f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:27:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c3a65b185990bb86e68d33bb150368dc6a74702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:06:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-containers-multinode/ba190f7/ : SUCCESS in 2h 06m 05s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-standalone/aa9b70e/ : SUCCESS in 2h 08m 35s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-fedora-28-standalone/bcc429e/ : FAILURE in 2h 04m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/35bec51/ : SUCCESS in 2h 11m 27s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/gate/openstack-tox-pep8/f9dc9ef/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py27/8fc04aa/ : SUCCESS in 6m 42s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py35/75ff391/ : SUCCESS in 5m 57s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py36/6286147/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/gate/openstack-tox-docs/a550818/html/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/gate/openstack-tox-lower-constraints/d41b6bc/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4941e48ba7515d29d8db1d9589ce138043272cc7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 19:02:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89c296e93e9324fae0cfdd51cdf28b11b3e96dc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-containers-multinode/2f0d22b/ : SUCCESS in 2h 12m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-undercloud-containers/042d4fd/ : SUCCESS in 1h 27m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/07cd05d/ : SUCCESS in 2h 25m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/faa891d/ : SUCCESS in 2h 39m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-centos-7-standalone/2d61283/ : SUCCESS in 1h 51m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/check/tripleo-ci-fedora-28-standalone/c564129/ : SUCCESS in 1h 38m 41s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/check/openstack-tox-pep8/715a5ce/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/check/openstack-tox-py27/8a2803b/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/check/openstack-tox-py35/a560a5d/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/check/openstack-tox-py36/9d9ddae/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/check/openstack-tox-docs/7e4baa2/html/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/check/openstack-tox-lower-constraints/fa9e256/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c4f6d8e9e246d9e0886f9ab94a410fa95bbc613a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:24:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfa340ae7f41eaac18c46cee649d1d23d44bcf11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-containers-multinode/358521d/ : SUCCESS in 1h 57m 18s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-standalone/dc0d0db/ : SUCCESS in 1h 54m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/18/663918/1/gate/tripleo-ci-fedora-28-standalone/785dcf7/ : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/18/663918/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/09f40aa/ : SUCCESS in 2h 30m 36s\n- openstack-tox-pep8 http://logs.openstack.org/18/663918/1/gate/openstack-tox-pep8/ab3dd3d/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py27/8a0bcda/ : SUCCESS in 5m 19s\n- openstack-tox-py35 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py35/5686e6c/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/18/663918/1/gate/openstack-tox-py36/17d891c/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/18/663918/1/gate/openstack-tox-docs/243cd27/html/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/663918/1/gate/openstack-tox-lower-constraints/fc3d870/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4c6dd877385b9c6e5e5338a7aaea84cc5ff88c8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8727fda27dc62400509feb14d82412e72756ee6e","revisions":{"8727fda27dc62400509feb14d82412e72756ee6e":{"kind":"REWORK","_number":1,"created":"2019-06-07 12:10:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/18/663918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/18/663918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/18/663918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/18/663918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/18/663918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/18/663918/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 12:08:11.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 12:08:11.000000000","tz":120},"subject":"Sync \u0027validations/callback_plugins/validation_output.py\u0027","message":"Sync \u0027validations/callback_plugins/validation_output.py\u0027\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the custom callback plug-in with the one used by\nthe validation framework.\n\nChange-Id: I6555d97051b4c0a1361e69e24ef849bf98828e0b\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8727fda27dc62400509feb14d82412e72756ee6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8727fda27dc62400509feb14d82412e72756ee6e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
