)]}'
{"id":"openstack%2Ftripleo-validations~663920","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ibd8548879d99902eae333f2bfc1e9ed3b2f43e86","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"Ibd8548879d99902eae333f2bfc1e9ed3b2f43e86","subject":"Sync \u0027validations/lookup_plugins/tht.py\u0027","status":"MERGED","created":"2019-06-07 12:26:16.000000000","updated":"2019-06-10 04:41:05.000000000","submitted":"2019-06-10 04:41:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663920-1560141665427-d9d96933","meta_rev_id":"2ec55f70c12c873aa2aad8e4fdd1b5b69d834b4b","_number":663920,"virtual_id_number":663920,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 04:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:20:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:28:32.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:34:06.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:20:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:28:32.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:28:32.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:28:32.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:20:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 04:41:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3805bfa5185a2e41cfc975aa55d1439546c11ab","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 12:26:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bb4b7cb2290d86d6a40bf25e8a436cb7f741190","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:28:57.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"818aa00496708fc7ee9a133b6df6a005f5604419","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:33:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/663920/1/check/tripleo-ci-centos-7-containers-multinode/02878f1/ : SUCCESS in 2h 05m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/663920/1/check/tripleo-ci-centos-7-undercloud-containers/e8609b8/ : SUCCESS in 1h 34m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/663920/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/21f99ab/ : SUCCESS in 2h 04m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/663920/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/29f9d27/ : SUCCESS in 2h 22m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/663920/1/check/tripleo-ci-centos-7-standalone/9f62a21/ : SUCCESS in 1h 45m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/663920/1/check/tripleo-ci-fedora-28-standalone/99361dc/ : SUCCESS in 1h 43m 30s\n- openstack-tox-pep8 http://logs.openstack.org/20/663920/1/check/openstack-tox-pep8/818c8b0/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/20/663920/1/check/openstack-tox-py27/d5a8205/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/20/663920/1/check/openstack-tox-py35/706084b/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/20/663920/1/check/openstack-tox-py36/0e1c440/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/20/663920/1/check/openstack-tox-docs/dca180f/html/ : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/663920/1/check/openstack-tox-lower-constraints/d9abfdd/ : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ede4f3c67d5211470cf2b5b76b893bc3a369f8eb","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:34:06.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"51874ce5c6d7e47ae441468ddd4a5e4d3b23940d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:20:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86c438f20e72fa97e4aca39685072c57ffaf94ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:20:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e204e5bf40ab0db1ef0ef1aee035fa7ac1c195c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 04:41:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b892c1ce5cb9a189181142a7271f57404a111ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 04:41:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/663920/1/gate/tripleo-ci-centos-7-containers-multinode/c0d3377/ : SUCCESS in 1h 58m 21s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/663920/1/gate/tripleo-ci-centos-7-standalone/dd5b775/ : SUCCESS in 1h 46m 43s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/663920/1/gate/tripleo-ci-fedora-28-standalone/7424da6/ : SUCCESS in 1h 55m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/663920/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/25238ac/ : SUCCESS in 2h 18m 00s\n- openstack-tox-pep8 http://logs.openstack.org/20/663920/1/gate/openstack-tox-pep8/5c5f0f9/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/20/663920/1/gate/openstack-tox-py27/3c2ccfc/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/20/663920/1/gate/openstack-tox-py35/f29fc0b/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/20/663920/1/gate/openstack-tox-py36/7160a13/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/20/663920/1/gate/openstack-tox-docs/a342e66/html/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/663920/1/gate/openstack-tox-lower-constraints/05aca7d/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97ee5d4bfe646313bb57a60014866db20d7d72ae","revisions":{"97ee5d4bfe646313bb57a60014866db20d7d72ae":{"kind":"REWORK","_number":1,"created":"2019-06-07 12:26:16.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/663920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/663920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/663920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/663920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/663920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/663920/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 12:24:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 12:24:28.000000000","tz":120},"subject":"Sync \u0027validations/lookup_plugins/tht.py\u0027","message":"Sync \u0027validations/lookup_plugins/tht.py\u0027\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the tht.py lookup plugin.\n\nChange-Id: Ibd8548879d99902eae333f2bfc1e9ed3b2f43e86\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/97ee5d4bfe646313bb57a60014866db20d7d72ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/97ee5d4bfe646313bb57a60014866db20d7d72ae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
