)]}'
{"id":"openstack%2Ftripleo-validations~669070","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I645af2be9a1c269ed4a635e3299164c94d6ac810","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"ceph-health_vars-stable/stein","hashtags":[],"change_id":"I645af2be9a1c269ed4a635e3299164c94d6ac810","subject":"Fix variables indentation for ceph-health validation","status":"MERGED","created":"2019-07-04 07:15:25.000000000","updated":"2019-07-08 10:17:29.000000000","submitted":"2019-07-08 10:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669070-1562581049026-d0a65de9","meta_rev_id":"93b10164b9a102a1ba0815af9fcbb9531d001e3a","_number":669070,"virtual_id_number":669070,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-07-08 10:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"recommended":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-04 10:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-04 07:42:20.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-04 07:15:53.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-07-04 07:15:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-04 07:15:34.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":1,"date":"2019-07-08 07:40:31.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 07:15:34.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-07-04 07:15:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-04 07:15:53.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-04 07:42:20.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-04 10:43:32.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-07-08 07:40:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-08 10:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d99cb325852b1a9a91bfea91be5c44133770f79","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-04 07:15:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2a7583f17c7b0a878c0ccf87ab8451d17312d638","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-04 07:42:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c56fffe7599a8b4066eaaaa7f94e3840d454640","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:13:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/669070/1/check/tripleo-ci-centos-7-containers-multinode/5551571/ : SUCCESS in 2h 25m 40s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/669070/1/check/tripleo-ci-centos-7-undercloud-containers/862bcbf/ : SUCCESS in 1h 41m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/669070/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2011255/ : SUCCESS in 2h 45m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/669070/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e392800/ : SUCCESS in 2h 31m 42s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/669070/1/check/tripleo-ci-centos-7-standalone/b057bf6/ : SUCCESS in 2h 03m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/669070/1/check/tripleo-ci-fedora-28-standalone/a637e1a/ : SUCCESS in 1h 45m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/669070/1/check/openstack-tox-pep8/5c5a04d/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/70/669070/1/check/openstack-tox-py27/a1ff5ab/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/70/669070/1/check/openstack-tox-py35/43ec148/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/70/669070/1/check/openstack-tox-py36/8b3d444/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/70/669070/1/check/openstack-tox-docs/38cad6b/html/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/669070/1/check/openstack-tox-lower-constraints/dd85fc9/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"afa31ca2331531357cf199a20a5e7dd0344aee35","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-07-04 10:43:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d919838712f5565acf261b16ff76d47ea97c3ff","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 07:40:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50a395a51a49ed81087d5dd2c8bb398c6d571a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 07:40:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db1932681d6860c4702180a6a37d8ab713d378bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:17:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/669070/1/gate/tripleo-ci-centos-7-containers-multinode/7b5d428/ : SUCCESS in 2h 17m 11s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/669070/1/gate/tripleo-ci-centos-7-standalone/f92c70f/ : SUCCESS in 1h 45m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/669070/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2fb1c02/ : SUCCESS in 2h 34m 54s\n- openstack-tox-pep8 http://logs.openstack.org/70/669070/1/gate/openstack-tox-pep8/486bce0/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/70/669070/1/gate/openstack-tox-py27/3c555e4/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/70/669070/1/gate/openstack-tox-py35/a6cea4d/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/70/669070/1/gate/openstack-tox-py36/18a274b/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/70/669070/1/gate/openstack-tox-docs/fe2d6ba/html/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/669070/1/gate/openstack-tox-lower-constraints/39b0bff/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"90a9d881e96bb3cc8df69f55d718def05505cebd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:17:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bdb3b674b4bf2bbf6dab1685c6abfd4a12f7af7","revisions":{"4bdb3b674b4bf2bbf6dab1685c6abfd4a12f7af7":{"kind":"REWORK","_number":1,"created":"2019-07-04 07:15:25.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/70/669070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/70/669070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/669070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/669070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/669070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/70/669070/1"}}},"commit":{"parents":[{"commit":"d4442605e6d79f4b09c73c60984a331937dd4092","subject":"Merge \"Remove useless cookiecutter files\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d4442605e6d79f4b09c73c60984a331937dd4092"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-03 11:09:39.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-04 07:15:25.000000000","tz":0},"subject":"Fix variables indentation for ceph-health validation","message":"Fix variables indentation for ceph-health validation\n\nThe default variables for a validation playbook should not be included\nin the metadata array level.\n\nChange-Id: I645af2be9a1c269ed4a635e3299164c94d6ac810\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit bb48a7f577ffe4d40b95c92b151122a9d581505b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4bdb3b674b4bf2bbf6dab1685c6abfd4a12f7af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4bdb3b674b4bf2bbf6dab1685c6abfd4a12f7af7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
