)]}'
{"id":"openstack%2Ftripleo-validations~669397","triplet_id":"openstack%2Ftripleo-validations~master~I5c08dc36c811c76d4b4c5ee6a0872ad031c243f8","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validations-framework","hashtags":[],"change_id":"I5c08dc36c811c76d4b4c5ee6a0872ad031c243f8","subject":"Remove useless meta directory for the Ansible ntp role","status":"MERGED","created":"2019-07-05 16:51:25.000000000","updated":"2019-08-12 09:32:51.000000000","submitted":"2019-08-12 09:32:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669397-1565602371619-53e3b4f1","meta_rev_id":"5f186636d0477bf302edc1f004eb2db6adbbd618","_number":669397,"virtual_id_number":669397,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-08-12 09:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-08-09 11:01:06.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-08-09 11:01:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":1,"date":"2019-08-09 11:01:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":1,"date":"2019-08-10 16:50:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-12 06:49:20.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 16:51:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 16:51:53.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-08 13:32:07.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-31 13:35:45.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-07-31 19:59:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-01 06:35:03.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-10 16:50:56.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-08-12 06:49:20.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-12 09:32:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56c9a189faafd336ec446a77ed7bee0b9d57f839","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-05 16:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64d7b92e276fcea916619f369d8292a831ed4fbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:04:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-containers-multinode/00a69e5/ : SUCCESS in 2h 34m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-undercloud-containers/a46632c/ : SUCCESS in 1h 26m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3f1c6fa/ : SUCCESS in 2h 23m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c69c8d2/ : SUCCESS in 2h 34m 13s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone/ffec0d7/ : SUCCESS in 2h 13m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669397/1/check/tripleo-ci-fedora-28-standalone/709afba/ : SUCCESS in 1h 37m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone-upgrade/4180124/ : SUCCESS in 2h 27m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/660137e/ : FAILURE in 1h 18m 56s\n- openstack-tox-pep8 http://logs.openstack.org/97/669397/1/check/openstack-tox-pep8/e6421fa/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/97/669397/1/check/openstack-tox-py27/883bdfa/ : SUCCESS in 5m 52s\n- openstack-tox-py35 http://logs.openstack.org/97/669397/1/check/openstack-tox-py35/4838075/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/97/669397/1/check/openstack-tox-py36/f33929b/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/97/669397/1/check/openstack-tox-docs/28b5f5b/html/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669397/1/check/openstack-tox-lower-constraints/b2af363/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"747231018a3743d30410600ce662a7a5df95184f","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-08 13:32:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d062318624095c442a99a8103ac8f0d1ccd636fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:47:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-containers-multinode/56fee44/ : SUCCESS in 2h 24m 20s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-undercloud-containers/9dd7270/ : SUCCESS in 1h 37m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6f55f35/ : SUCCESS in 2h 18m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ed3aedf/ : SUCCESS in 2h 40m 47s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone/ad7b5e3/ : SUCCESS in 2h 04m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669397/1/check/tripleo-ci-fedora-28-standalone/230ce48/ : FAILURE in 28m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone-upgrade/a5fe9d8/ : SUCCESS in 2h 10m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/97/669397/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/93f774c/ : SUCCESS in 2h 12m 39s\n- openstack-tox-pep8 http://logs.openstack.org/97/669397/1/check/openstack-tox-pep8/cf8c13a/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/97/669397/1/check/openstack-tox-py27/441a528/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/97/669397/1/check/openstack-tox-py35/0611a1f/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/97/669397/1/check/openstack-tox-py36/f94647b/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/97/669397/1/check/openstack-tox-docs/5aee6a5/html/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669397/1/check/openstack-tox-lower-constraints/439ca7a/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"89789ce3f87ac25ac656366728ea17c9a170a566","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 13:23:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b7c90b257df9218f9068ace50a0cc4dee6e35008","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-31 13:35:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6167e3609fe99013aa6630a8176b4923178f6ea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 19:56:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-containers-multinode/5eeb6a0/ : SUCCESS in 2h 12m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-undercloud-containers/87ab301/ : SUCCESS in 1h 45m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1bb71f0/ : SUCCESS in 2h 37m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/25bc826/ : SUCCESS in 2h 53m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-standalone/09fa8a8/ : SUCCESS in 1h 53m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669397/2/check/tripleo-ci-fedora-28-standalone/7df7068/ : SUCCESS in 1h 47m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669397/2/check/tripleo-ci-centos-7-standalone-upgrade/3ca1829/ : SUCCESS in 2h 04m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/669397/2/check/openstack-tox-pep8/64dc9d6/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/97/669397/2/check/openstack-tox-py27/bf3059e/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/97/669397/2/check/openstack-tox-py36/c0fde33/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/97/669397/2/check/openstack-tox-py37/303ca0c/ : SUCCESS in 5m 01s\n- openstack-tox-molecule http://logs.openstack.org/97/669397/2/check/openstack-tox-molecule/397be17/tox/reports.html : SUCCESS in 30m 59s\n- openstack-tox-docs http://logs.openstack.org/97/669397/2/check/openstack-tox-docs/b5ec2ee/html/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669397/2/check/openstack-tox-lower-constraints/8878872/ : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"92c3213e41a672b4cc9108ac21a8ac48afdf43b0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-31 19:59:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6842e6dcda673f1953adf9777c6969b73db4cd7","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-01 06:35:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9369c6bf8815d6c6595b11c3fdbed7661d96b6fb","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-05 11:08:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c52b4f0d4cfb12e05c41569b69bed1493064eb55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 11:08:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26773db9853e49303e6c3e6f98434fdcba715666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:29:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/97/669397/2/gate/tripleo-ci-centos-7-containers-multinode/0b52c33/ : TIMED_OUT in 3h 19m 02s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/97/669397/2/gate/tripleo-ci-centos-7-standalone/efe163e/ : SUCCESS in 2h 07m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/97/669397/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/44e0a9b/ : SUCCESS in 2h 19m 47s\n- openstack-tox-pep8 https://logs.opendev.org/97/669397/2/gate/openstack-tox-pep8/35b0edc/ : SUCCESS in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/97/669397/2/gate/openstack-tox-py27/bc238e9/ : SUCCESS in 3m 17s\n- openstack-tox-py36 https://logs.opendev.org/97/669397/2/gate/openstack-tox-py36/67d58ab/ : SUCCESS in 4m 13s\n- openstack-tox-py37 https://logs.opendev.org/97/669397/2/gate/openstack-tox-py37/e875f6d/ : SUCCESS in 4m 12s\n- openstack-tox-molecule https://logs.opendev.org/97/669397/2/gate/openstack-tox-molecule/e6c46c7/tox/reports.html : SUCCESS in 29m 40s\n- openstack-tox-docs https://logs.opendev.org/97/669397/2/gate/openstack-tox-docs/568d116/html/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/97/669397/2/gate/openstack-tox-lower-constraints/96c779a/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"46a12add8d6e363883cd2c141e8ac8b15aa1e249","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 13:13:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac6b9e8ca512d7cdec9129229c650537e7b58347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:27:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-containers-multinode/54bf019/ : SUCCESS in 2h 40m 52s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-undercloud-containers/82e6588/ : SUCCESS in 1h 49m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2f09b90/ : FAILURE in 1h 17m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/75dfb01/ : SUCCESS in 2h 52m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-standalone/b782af4/ : SUCCESS in 2h 09m 23s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/97/669397/2/check/tripleo-ci-fedora-28-standalone/f0327e3/ : SUCCESS in 1h 43m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/97/669397/2/check/tripleo-ci-centos-7-standalone-upgrade/6bfd339/ : SUCCESS in 2h 19m 32s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/97/669397/2/check/openstack-tox-pep8/2f3fe2d/ : SUCCESS in 5m 58s\n- openstack-tox-py27 https://logs.opendev.org/97/669397/2/check/openstack-tox-py27/1814d1d/ : SUCCESS in 6m 03s\n- openstack-tox-py36 https://logs.opendev.org/97/669397/2/check/openstack-tox-py36/31c8da9/ : SUCCESS in 4m 38s\n- openstack-tox-py37 https://logs.opendev.org/97/669397/2/check/openstack-tox-py37/2bd0048/ : SUCCESS in 8m 26s\n- openstack-tox-molecule https://logs.opendev.org/97/669397/2/check/openstack-tox-molecule/623baa7/tox/reports.html : TIMED_OUT in 31m 19s\n- openstack-tox-docs https://logs.opendev.org/97/669397/2/check/openstack-tox-docs/710b208/html/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://logs.opendev.org/97/669397/2/check/openstack-tox-lower-constraints/adc2821/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b5fc355f85371f7d9f31aab085619e977e000ff5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-09 11:01:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"511f85683a3e2812c40aad5049adfc78053dfbfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 16:29:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-containers-multinode/1826fd3/ : SUCCESS in 2h 13m 34s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-undercloud-containers/da1353d/ : SUCCESS in 1h 26m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e54c1f1/ : SUCCESS in 2h 56m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/40530e7/ : SUCCESS in 2h 18m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-standalone/92cb9d2/ : SUCCESS in 1h 52m 01s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/97/669397/3/check/tripleo-ci-fedora-28-standalone/c9181ee/ : SUCCESS in 1h 46m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/97/669397/3/check/tripleo-ci-centos-7-standalone-upgrade/5506134/ : SUCCESS in 2h 14m 49s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/97/669397/3/check/openstack-tox-pep8/2dfbbba/ : SUCCESS in 4m 14s\n- openstack-tox-py27 https://logs.opendev.org/97/669397/3/check/openstack-tox-py27/7ef0872/ : SUCCESS in 4m 54s\n- openstack-tox-py36 https://logs.opendev.org/97/669397/3/check/openstack-tox-py36/1f14f10/ : SUCCESS in 4m 08s\n- openstack-tox-py37 https://logs.opendev.org/97/669397/3/check/openstack-tox-py37/67550ce/ : SUCCESS in 4m 49s\n- openstack-tox-docs https://logs.opendev.org/97/669397/3/check/openstack-tox-docs/3c07c6e/html/ : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://logs.opendev.org/97/669397/3/check/openstack-tox-lower-constraints/175d89f/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"cc93ede338e6452fbeb56a150b0415accc6d3cc6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-08-10 16:50:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"24f3e94c0d832bdf4a6f324a0985ad95cd717e60","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-12 06:49:20.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6277edb73100f2f9a1ec01cad70dd718f2fcd7b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 06:49:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e52c1fe0bc6d5fdb9dfba72ae0e26f3379333ff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:32:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e78f574f32e3f3afd1a1fbb485054776fe16f777","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:32:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/97/669397/3/gate/tripleo-ci-centos-7-containers-multinode/90dde5d/ : SUCCESS in 2h 22m 28s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/97/669397/3/gate/tripleo-ci-centos-7-standalone/9c651f6/ : SUCCESS in 1h 44m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/97/669397/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1e77c81/ : SUCCESS in 2h 03m 32s\n- openstack-tox-pep8 https://logs.opendev.org/97/669397/3/gate/openstack-tox-pep8/ff34229/ : SUCCESS in 4m 12s\n- openstack-tox-py27 https://logs.opendev.org/97/669397/3/gate/openstack-tox-py27/71c748e/ : SUCCESS in 3m 39s\n- openstack-tox-py36 https://logs.opendev.org/97/669397/3/gate/openstack-tox-py36/9a69d7a/ : SUCCESS in 5m 32s\n- openstack-tox-py37 https://logs.opendev.org/97/669397/3/gate/openstack-tox-py37/0b93c45/ : SUCCESS in 6m 32s\n- openstack-tox-docs https://logs.opendev.org/97/669397/3/gate/openstack-tox-docs/07b678a/html/ : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://logs.opendev.org/97/669397/3/gate/openstack-tox-lower-constraints/1faace7/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3eb813fee773eccee7e5155a22c141df8b93c0f","revisions":{"954351d556e4328ea8b4cb444c6c84e1a6cc24aa":{"kind":"REWORK","_number":1,"created":"2019-07-05 16:51:25.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/97/669397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/97/669397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/1"}}},"commit":{"parents":[{"commit":"bb48a7f577ffe4d40b95c92b151122a9d581505b","subject":"Fix variables indentation for ceph-health validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bb48a7f577ffe4d40b95c92b151122a9d581505b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-05 16:50:00.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-05 16:50:00.000000000","tz":120},"subject":"Remove useless meta directory for the Ansible ntp role","message":"Remove useless meta directory for the Ansible ntp role\n\nChange-Id: I5c08dc36c811c76d4b4c5ee6a0872ad031c243f8\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/954351d556e4328ea8b4cb444c6c84e1a6cc24aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/954351d556e4328ea8b4cb444c6c84e1a6cc24aa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bb48a7f577ffe4d40b95c92b151122a9d581505b","is_merged_in_target_branch":true,"change_id":"I645af2be9a1c269ed4a635e3299164c94d6ac810","change_number":668866,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8f71af668e5df4d78b1f03416de1790ce80fd60c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-31 13:23:32.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/97/669397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/97/669397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/2"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-05 16:50:00.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 13:23:32.000000000","tz":0},"subject":"Remove useless meta directory for the Ansible ntp role","message":"Remove useless meta directory for the Ansible ntp role\n\nChange-Id: I5c08dc36c811c76d4b4c5ee6a0872ad031c243f8\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8f71af668e5df4d78b1f03416de1790ce80fd60c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8f71af668e5df4d78b1f03416de1790ce80fd60c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e3eb813fee773eccee7e5155a22c141df8b93c0f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-09 11:01:06.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/97/669397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/97/669397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/97/669397/3"}}},"commit":{"parents":[{"commit":"741b36362dc3341feb06f15c2b89a814c25b1308","subject":"Merge \"Use custom Ceph cluster name in validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/741b36362dc3341feb06f15c2b89a814c25b1308"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-05 16:50:00.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-09 11:01:06.000000000","tz":0},"subject":"Remove useless meta directory for the Ansible ntp role","message":"Remove useless meta directory for the Ansible ntp role\n\nChange-Id: I5c08dc36c811c76d4b4c5ee6a0872ad031c243f8\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e3eb813fee773eccee7e5155a22c141df8b93c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e3eb813fee773eccee7e5155a22c141df8b93c0f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"741b36362dc3341feb06f15c2b89a814c25b1308","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
