)]}'
{"id":"openstack%2Ftripleo-validations~670833","triplet_id":"openstack%2Ftripleo-validations~master~I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c","project":"openstack/tripleo-validations","branch":"master","topic":"molecule","hashtags":[],"change_id":"I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c","subject":"Make Verify the RAM requirements task failing again","status":"MERGED","created":"2019-07-15 13:33:36.000000000","updated":"2019-07-16 10:48:44.000000000","submitted":"2019-07-16 10:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670833-1563274124160-77280bbb","meta_rev_id":"09c990098b1363a2016916a018c697f8b1459152","_number":670833,"virtual_id_number":670833,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-16 10:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 05:19:28.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-16 08:19:35.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 13:34:39.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-15 13:34:48.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-15 13:34:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-15 13:34:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-16 08:19:35.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-16 10:48:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3dd610bcb9d6f44236211be759bdcd017d1d6cc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 13:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318aaa381e8e6a1cc309e778118f4a94eca0f6ac","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 14:04:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7479faf390236047a2cc1d91083795cd68c065fc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 14:07:52.000000000","message":"Topic set to molecule","accounts_in_message":[],"_revision_number":2},{"id":"578f33c50b6abd94d77b4f8986acea7bca047caf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:24:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-containers-multinode/ec15a92/ : SUCCESS in 2h 20m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-undercloud-containers/683c784/ : SUCCESS in 1h 35m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/dfe4b4a/ : SUCCESS in 2h 25m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b048330/ : FAILURE in 2h 51m 50s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-standalone/4b2bdee/ : SUCCESS in 1h 51m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/33/670833/2/check/tripleo-ci-fedora-28-standalone/4e8b6b2/ : SUCCESS in 1h 52m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/33/670833/2/check/tripleo-ci-centos-7-standalone-upgrade/be10be6/ : FAILURE in 2h 22m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/33/670833/2/check/openstack-tox-pep8/05ad7cc/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/33/670833/2/check/openstack-tox-py27/72c3dcd/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/33/670833/2/check/openstack-tox-py36/2b20b9b/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/33/670833/2/check/openstack-tox-py37/fd5e182/ : SUCCESS in 8m 57s\n- openstack-tox-molecule http://logs.openstack.org/33/670833/2/check/openstack-tox-molecule/255acc9/tox/reports.html : SUCCESS in 12m 11s\n- openstack-tox-docs http://logs.openstack.org/33/670833/2/check/openstack-tox-docs/a26530a/html/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/670833/2/check/openstack-tox-lower-constraints/f9838c2/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5c84e1c74246194131be8350bb408ab610e6d178","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-16 05:19:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"334059bfbfc495b49df62d4adf47e99b5c609088","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-16 08:19:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2058d1568c58f165c30fee333e0f512b6504fb96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 08:19:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7da94de9f105581995e0b5ff15b701eaee4565f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:48:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/33/670833/2/gate/tripleo-ci-centos-7-containers-multinode/96eae65/ : SUCCESS in 2h 08m 51s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/33/670833/2/gate/tripleo-ci-centos-7-standalone/75ea43b/ : SUCCESS in 2h 05m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/33/670833/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7db7f0a/ : SUCCESS in 2h 27m 19s\n- openstack-tox-pep8 http://logs.openstack.org/33/670833/2/gate/openstack-tox-pep8/0d796b7/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/33/670833/2/gate/openstack-tox-py27/25109b2/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/33/670833/2/gate/openstack-tox-py36/14af869/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/33/670833/2/gate/openstack-tox-py37/be05490/ : SUCCESS in 5m 32s\n- openstack-tox-molecule http://logs.openstack.org/33/670833/2/gate/openstack-tox-molecule/1bbefbb/tox/reports.html : SUCCESS in 10m 17s\n- openstack-tox-docs http://logs.openstack.org/33/670833/2/gate/openstack-tox-docs/364ca6c/html/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/670833/2/gate/openstack-tox-lower-constraints/c5cf369/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"09c990098b1363a2016916a018c697f8b1459152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:48:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e6c26df546ee24903352f66d4a2e5d95fff10ad","revisions":{"4b7855dd4e135893d9ad101116c2dcf48539b978":{"kind":"REWORK","_number":1,"created":"2019-07-15 13:33:36.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/33/670833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/33/670833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/1"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:29:19.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:29:19.000000000","tz":120},"subject":"Make Verify the RAM requirements task failing again","message":"Make Verify the RAM requirements task failing again\n\nCommit 5bd41653b6bf097bc505b56ff539c26e440cdadb changes the tasks to use\nthe debug module instead of fail module.\n\nThis patch makes that task failing again.\n\nChange-Id: I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4b7855dd4e135893d9ad101116c2dcf48539b978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4b7855dd4e135893d9ad101116c2dcf48539b978"}]},"branch":"refs/heads/master"},"4e6c26df546ee24903352f66d4a2e5d95fff10ad":{"kind":"REWORK","_number":2,"created":"2019-07-15 14:04:38.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/33/670833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/33/670833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/33/670833/2"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:29:19.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 14:03:37.000000000","tz":120},"subject":"Make Verify the RAM requirements task failing again","message":"Make Verify the RAM requirements task failing again\n\nCommit 5bd41653b6bf097bc505b56ff539c26e440cdadb changes the tasks to use\nthe debug module instead of fail module.\n\nThis patch makes that task failing again and removes the useless\nverify.yml file.\n\nChange-Id: I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4e6c26df546ee24903352f66d4a2e5d95fff10ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4e6c26df546ee24903352f66d4a2e5d95fff10ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
