)]}'
{"id":"openstack%2Ftripleo-validations~670835","triplet_id":"openstack%2Ftripleo-validations~master~I1951ea46bd426b55157d6b1b863fd8a46c110755","project":"openstack/tripleo-validations","branch":"master","topic":"molecule","hashtags":[],"change_id":"I1951ea46bd426b55157d6b1b863fd8a46c110755","subject":"Add molecule tests for the undercloud-cpu role","status":"MERGED","created":"2019-07-15 13:43:20.000000000","updated":"2019-07-16 20:41:08.000000000","submitted":"2019-07-16 20:41:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670835-1563309668638-14776782","meta_rev_id":"9e95578c15f3ba8ebb3e1fff158f6cd14bd70645","_number":670835,"virtual_id_number":670835,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-16 20:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"recommended":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-07-15 17:05:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 05:18:00.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-07-16 13:21:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 13:44:13.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-15 17:05:15.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-16 05:18:00.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-16 13:21:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-16 20:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71dc044a88fc7a7e7f0f529418031dad26875a0c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 13:43:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f904db0bbefc0205493c14f7ddb1ed42fc3cc1f","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-15 13:49:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6de0033c6b95e17f4237edd30041d783c448169a","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-15 13:50:42.000000000","message":"Patch Set 1:\n\nWe should also set the topic to \"molecule\" to help keep track of these tests we\u0027re adding.","accounts_in_message":[],"_revision_number":1},{"id":"306bc1ce826e3c013d72042e436e32e853ff5eeb","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 14:00:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f12497ed23dbd60375c70985ba756ea3672cad5","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-15 17:05:15.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for the update, looks great!","accounts_in_message":[],"_revision_number":2},{"id":"6f0ec29cfdf9b01222d8b94eaf6167656bd0bfc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:11:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-containers-multinode/72433ad/ : SUCCESS in 2h 27m 26s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-undercloud-containers/400359e/ : SUCCESS in 1h 43m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1cb5c73/ : SUCCESS in 2h 43m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f641353/ : FAILURE in 2h 25m 44s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-standalone/250a049/ : SUCCESS in 3h 13m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670835/2/check/tripleo-ci-fedora-28-standalone/7f27b53/ : SUCCESS in 1h 45m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670835/2/check/tripleo-ci-centos-7-standalone-upgrade/14f450a/ : FAILURE in 2h 02m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/670835/2/check/openstack-tox-pep8/76f5a09/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/35/670835/2/check/openstack-tox-py27/e5c445d/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/35/670835/2/check/openstack-tox-py36/9e2a313/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/35/670835/2/check/openstack-tox-py37/bd50de2/ : SUCCESS in 6m 08s\n- openstack-tox-molecule http://logs.openstack.org/35/670835/2/check/openstack-tox-molecule/ec9bb39/tox/reports.html : SUCCESS in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/35/670835/2/check/openstack-tox-docs/4b324ed/html/ : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/670835/2/check/openstack-tox-lower-constraints/d325a8b/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2df6cc2343f474eb97cdbf39f4bba5cdec136ac6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-16 05:18:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3d534ce289cd9f8dedd2298dea7a0b4e8ae0bf89","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-16 13:21:12.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fe2d0d93b27c2db100fde462af3aa15a544344bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 13:21:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6477de3bfe96fc9e23e85d821c36f2fb8e840486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"68cd49d40386ec2444bb36a8d2812a82b35cb353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:41:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670835/2/gate/tripleo-ci-centos-7-containers-multinode/16d4fb3/ : SUCCESS in 2h 36m 26s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670835/2/gate/tripleo-ci-centos-7-standalone/9899cce/ : SUCCESS in 1h 55m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670835/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6b1de94/ : SUCCESS in 2h 22m 39s\n- openstack-tox-pep8 http://logs.openstack.org/35/670835/2/gate/openstack-tox-pep8/3c8db1b/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/35/670835/2/gate/openstack-tox-py27/7e219b5/ : SUCCESS in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/35/670835/2/gate/openstack-tox-py36/64f6f33/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/35/670835/2/gate/openstack-tox-py37/dbfe7dc/ : SUCCESS in 5m 23s\n- openstack-tox-molecule http://logs.openstack.org/35/670835/2/gate/openstack-tox-molecule/f46de8b/tox/reports.html : SUCCESS in 12m 57s\n- openstack-tox-docs http://logs.openstack.org/35/670835/2/gate/openstack-tox-docs/1877ad8/html/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/670835/2/gate/openstack-tox-lower-constraints/25960f7/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f0e39a7febd3fc5bc7bb0b20250d4280652bc6b","revisions":{"9b46e8591d5544ae53e62456e5cec42c7162f7f6":{"kind":"REWORK","_number":1,"created":"2019-07-15 13:43:20.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/35/670835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/35/670835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/1"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:37:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:37:53.000000000","tz":120},"subject":"Add molecule tests for the undercloud-cpu role","message":"Add molecule tests for the undercloud-cpu role\n\nChange-Id: I1951ea46bd426b55157d6b1b863fd8a46c110755\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b46e8591d5544ae53e62456e5cec42c7162f7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b46e8591d5544ae53e62456e5cec42c7162f7f6"}]},"branch":"refs/heads/master"},"5f0e39a7febd3fc5bc7bb0b20250d4280652bc6b":{"kind":"REWORK","_number":2,"created":"2019-07-15 14:00:45.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/35/670835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/35/670835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/35/670835/2"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:37:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:58:45.000000000","tz":120},"subject":"Add molecule tests for the undercloud-cpu role","message":"Add molecule tests for the undercloud-cpu role\n\nChange-Id: I1951ea46bd426b55157d6b1b863fd8a46c110755\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5f0e39a7febd3fc5bc7bb0b20250d4280652bc6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5f0e39a7febd3fc5bc7bb0b20250d4280652bc6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
