)]}'
{"id":"openstack%2Ftripleo-validations~671020","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule-stable-stein","hashtags":[],"change_id":"I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c","subject":"Make Verify the RAM requirements task failing again","status":"MERGED","created":"2019-07-16 12:43:46.000000000","updated":"2019-08-02 01:15:34.000000000","submitted":"2019-08-02 01:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671020-1564708534960-bfcf0b41","meta_rev_id":"436d4a2abaa5e9c0541861f71a6d92840d3d8fba","_number":671020,"virtual_id_number":671020,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 01:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-08-01 22:37:05.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-01 22:37:05.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 12:44:10.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-16 12:44:13.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-16 12:44:25.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-01 22:37:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 01:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8ea68a1b9f175a96ed4a4903838dbf7d5cd7ae6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-16 12:43:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37085f4a4a31ab611e9a0c7ee7af39d0ab0ca4ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:15:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/671020/1/check/tripleo-ci-centos-7-containers-multinode/293a238/ : FAILURE in 2h 13m 59s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/671020/1/check/tripleo-ci-centos-7-undercloud-containers/e343720/ : SUCCESS in 1h 44m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/671020/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/45aba51/ : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/671020/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/baa3aa1/ : SUCCESS in 2h 46m 56s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/671020/1/check/tripleo-ci-centos-7-standalone/9b75b5c/ : SUCCESS in 2h 03m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/671020/1/check/tripleo-ci-fedora-28-standalone/116cf71/ : SUCCESS in 1h 36m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/671020/1/check/openstack-tox-pep8/0af4c28/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/20/671020/1/check/openstack-tox-py27/6208c0c/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/20/671020/1/check/openstack-tox-py35/3d583fb/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/20/671020/1/check/openstack-tox-py36/4165c59/ : SUCCESS in 3m 45s\n- openstack-tox-molecule http://logs.openstack.org/20/671020/1/check/openstack-tox-molecule/0e5a14b/tox/reports.html : FAILURE in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/20/671020/1/check/openstack-tox-docs/c3bcd83/html/ : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/671020/1/check/openstack-tox-lower-constraints/adc5394/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c46e73bd948e1b69fa88d64e9b2e53502c8b8872","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:48:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"da593d53d87623944bf3f9605dc08df365a4ab2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:54:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/20/671020/2/check/tripleo-ci-centos-7-containers-multinode/b507e03/ : SUCCESS in 2h 10m 24s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/20/671020/2/check/tripleo-ci-centos-7-undercloud-containers/e79850b/ : SUCCESS in 2h 17m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/20/671020/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f6d6852/ : SUCCESS in 2h 22m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/20/671020/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8ebe2c1/ : SUCCESS in 2h 41m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/20/671020/2/check/tripleo-ci-centos-7-standalone/6817ed3/ : SUCCESS in 1h 59m 23s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/20/671020/2/check/tripleo-ci-fedora-28-standalone/065a7e6/ : SUCCESS in 1h 56m 29s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/20/671020/2/check/openstack-tox-pep8/a64e17b/ : SUCCESS in 5m 31s\n- openstack-tox-py27 https://logs.opendev.org/20/671020/2/check/openstack-tox-py27/8736c97/ : SUCCESS in 3m 50s\n- openstack-tox-py35 https://logs.opendev.org/20/671020/2/check/openstack-tox-py35/7cbf8b0/ : SUCCESS in 4m 32s\n- openstack-tox-py36 https://logs.opendev.org/20/671020/2/check/openstack-tox-py36/aad209d/ : SUCCESS in 4m 41s\n- openstack-tox-molecule https://logs.opendev.org/20/671020/2/check/openstack-tox-molecule/ed9f850/tox/reports.html : SUCCESS in 13m 05s\n- openstack-tox-docs https://logs.opendev.org/20/671020/2/check/openstack-tox-docs/85444e6/html/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/671020/2/check/openstack-tox-lower-constraints/aaf85eb/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"005dbc92e6f88cb436b218f88732fbea2b9bd607","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 22:37:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c59d1be860e384aafd11ef170ce198531bbb822b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:37:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45e8fca49588045372119e6ceffbb76827cd52e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:15:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/20/671020/2/gate/tripleo-ci-centos-7-containers-multinode/767a9b2/ : SUCCESS in 2h 11m 34s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/20/671020/2/gate/tripleo-ci-centos-7-standalone/8e71aa9/ : SUCCESS in 1h 53m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/20/671020/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d146efa/ : SUCCESS in 2h 20m 43s\n- openstack-tox-pep8 https://logs.opendev.org/20/671020/2/gate/openstack-tox-pep8/b6ee4bd/ : SUCCESS in 4m 21s\n- openstack-tox-py27 https://logs.opendev.org/20/671020/2/gate/openstack-tox-py27/1cd0d1b/ : SUCCESS in 3m 47s\n- openstack-tox-py35 https://logs.opendev.org/20/671020/2/gate/openstack-tox-py35/09d4be3/ : SUCCESS in 4m 20s\n- openstack-tox-py36 https://logs.opendev.org/20/671020/2/gate/openstack-tox-py36/80d8789/ : SUCCESS in 5m 19s\n- openstack-tox-molecule https://logs.opendev.org/20/671020/2/gate/openstack-tox-molecule/df065a1/tox/reports.html : SUCCESS in 9m 38s\n- openstack-tox-docs https://logs.opendev.org/20/671020/2/gate/openstack-tox-docs/d8cfecb/html/ : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/671020/2/gate/openstack-tox-lower-constraints/73b3804/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"08d6a5c9dbf6c2aa54cc151e2054cdc6653fb65c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:15:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23dd988ac3c85891f3b2998d3625952e46b79908","revisions":{"5a0c5cdfd9e076c7d6d4eaa67287de23ded6756b":{"kind":"REWORK","_number":1,"created":"2019-07-16 12:43:46.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/671020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/671020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/1"}}},"commit":{"parents":[{"commit":"f954a927eb1e44e5c4d3b5e3636a954cf226ceea","subject":"Add Molecule tests for the role xfs-check-ftype.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f954a927eb1e44e5c4d3b5e3636a954cf226ceea"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:29:19.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-16 12:43:28.000000000","tz":120},"subject":"Make Verify the RAM requirements task failing again","message":"Make Verify the RAM requirements task failing again\n\nCommit 5bd41653b6bf097bc505b56ff539c26e440cdadb changes the tasks to use\nthe debug module instead of fail module.\n\nThis patch makes that task failing again and removes the useless\nverify.yml file.\n\nChange-Id: I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 4e6c26df546ee24903352f66d4a2e5d95fff10ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5a0c5cdfd9e076c7d6d4eaa67287de23ded6756b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5a0c5cdfd9e076c7d6d4eaa67287de23ded6756b"}]},"branch":"refs/heads/stable/stein"},"23dd988ac3c85891f3b2998d3625952e46b79908":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-01 12:48:10.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/671020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/671020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/671020/2"}}},"commit":{"parents":[{"commit":"c7923e06ae367b280ad290bf763bec347f9173ef","subject":"Add Molecule tests for the role xfs-check-ftype.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c7923e06ae367b280ad290bf763bec347f9173ef"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-15 13:29:19.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:47:38.000000000","tz":120},"subject":"Make Verify the RAM requirements task failing again","message":"Make Verify the RAM requirements task failing again\n\nCommit 5bd41653b6bf097bc505b56ff539c26e440cdadb changes the tasks to use\nthe debug module instead of fail module.\n\nThis patch makes that task failing again and removes the useless\nverify.yml file.\n\nChange-Id: I2fc48005c91dd5e4cf81a3b3eb2f1f73a2bca92c\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 4e6c26df546ee24903352f66d4a2e5d95fff10ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23dd988ac3c85891f3b2998d3625952e46b79908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23dd988ac3c85891f3b2998d3625952e46b79908"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
