)]}'
{"id":"openstack%2Ftripleo-validations~671460","triplet_id":"openstack%2Ftripleo-validations~master~Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed","project":"openstack/tripleo-validations","branch":"master","topic":"optimization/no-fact-gathering","hashtags":[],"change_id":"Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed","subject":"Enable fact gathering for needed validations","status":"MERGED","created":"2019-07-18 11:26:44.000000000","updated":"2019-08-01 10:34:06.000000000","submitted":"2019-07-31 17:40:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671460-1564594836347-b796e31c","meta_rev_id":"1d01eb0ecf99553d7da08c4781b6f502356c26c1","_number":671460,"virtual_id_number":671460,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-07-31 17:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-23 14:14:01.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-26 07:59:28.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-26 06:32:49.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"date":"2019-07-26 06:04:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-26 07:59:28.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"date":"2019-07-26 06:04:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 13:29:18.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-07-26 06:04:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-26 06:04:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-26 06:32:49.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-07-26 07:59:28.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-31 17:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a39f476c3ade0a8e3f69ca0171456b0ba30781f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-18 11:26:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a57eb804cc4488232fd4ac146471099d19d834","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-18 12:08:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nsee comments","accounts_in_message":[],"_revision_number":1},{"id":"2ac86523d70f699345d786a43218cf5a0c8a9dca","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-18 12:20:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"637705b22d34ab2b92ce3e69a820e3a65b436082","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-18 12:22:24.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ea3dd99e44de7e5b4d26ed83a4d4db0114e7314","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:05:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-containers-multinode/e995e40/ : SUCCESS in 2h 24m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-undercloud-containers/ae19817/ : SUCCESS in 1h 36m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f38a312/ : SUCCESS in 2h 15m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/68c0c86/ : FAILURE in 2h 40m 22s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-standalone/0fbbdfe/ : SUCCESS in 1h 47m 43s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/60/671460/2/check/tripleo-ci-fedora-28-standalone/c4697fc/ : SUCCESS in 1h 43m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/60/671460/2/check/tripleo-ci-centos-7-standalone-upgrade/47a086b/ : FAILURE in 1h 56m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/671460/2/check/openstack-tox-pep8/710a440/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/60/671460/2/check/openstack-tox-py27/7c746bb/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/60/671460/2/check/openstack-tox-py36/47926f7/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/60/671460/2/check/openstack-tox-py37/8d1f7e2/ : SUCCESS in 5m 32s\n- openstack-tox-molecule http://logs.openstack.org/60/671460/2/check/openstack-tox-molecule/0d99cc3/tox/reports.html : SUCCESS in 19m 57s\n- openstack-tox-docs http://logs.openstack.org/60/671460/2/check/openstack-tox-docs/9698025/html/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/671460/2/check/openstack-tox-lower-constraints/1b7b126/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a40909be16a7d2964a4102796f5d83234e5edf59","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-22 07:48:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d2c9bcb2621de682c4a4efedcd5c926f9acb49d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 10:32:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-containers-multinode/a4e4624/ : SUCCESS in 2h 20m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-undercloud-containers/f39c588/ : SUCCESS in 1h 29m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f4808c8/ : SUCCESS in 2h 09m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/25f1623/ : FAILURE in 2h 38m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-standalone/8ab3c1c/ : SUCCESS in 1h 54m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/60/671460/3/check/tripleo-ci-fedora-28-standalone/569d4af/ : SUCCESS in 1h 48m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/60/671460/3/check/tripleo-ci-centos-7-standalone-upgrade/bf3d230/ : FAILURE in 2h 04m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/671460/3/check/openstack-tox-pep8/5ab6b6c/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/60/671460/3/check/openstack-tox-py27/b9c308e/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/60/671460/3/check/openstack-tox-py36/45c9ab3/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/60/671460/3/check/openstack-tox-py37/d8d2784/ : SUCCESS in 4m 28s\n- openstack-tox-molecule http://logs.openstack.org/60/671460/3/check/openstack-tox-molecule/adca7ac/tox/reports.html : SUCCESS in 9m 55s\n- openstack-tox-docs http://logs.openstack.org/60/671460/3/check/openstack-tox-docs/a00ec47/html/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/671460/3/check/openstack-tox-lower-constraints/cc62768/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d38385ed517750b32ea71b6c8d7ef6c6625ee646","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-07-22 15:17:05.000000000","message":"Patch Set 3:\n\nmight as well add the change back into ironic-boot-configuration, right?","accounts_in_message":[],"_revision_number":3},{"id":"78c93bf8f95cafa00245449adbaf4b8c84a768fc","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 12:38:35.000000000","message":"Patch Set 3: Workflow-1\n\n@Jeremy: nope, we\u0027re make a change directly in tripleoclient and friends, so that by default it doesn\u0027t gather the facts. That will make things better. I\u0027ll update this change once we have proper things set up.","accounts_in_message":[],"_revision_number":3},{"id":"4d49b4ca3082dc3356fe8635ae2ef1a5c6b10849","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 13:07:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73011194751e3ba1856a0ba4c0dd01a4143d2461","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 13:09:21.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1272d54c52eaaf30f8afe010c3e7f5240e7badd0","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-07-23 13:29:18.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a5ac30038661cb16c920fa08bcc3290524b9674","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 13:58:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a5a8d1c38808e8e35034386bd22fc7806f237f9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:02:00.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"55341e5d312382b5a429e85e9255a179a8cddf31","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 14:14:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f9e51f88d4b544ea36589fd7f9387a21b5d08c23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:22:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-containers-multinode/18b2b77/ : SUCCESS in 2h 31m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-undercloud-containers/2f5ec51/ : SUCCESS in 1h 31m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0b30acf/ : SUCCESS in 3h 00m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/aa44375/ : SUCCESS in 2h 46m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-standalone/c810174/ : SUCCESS in 1h 48m 56s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/60/671460/6/check/tripleo-ci-fedora-28-standalone/716a73e/ : SUCCESS in 1h 50m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-standalone-upgrade/e0b7c43/ : FAILURE in 2h 10m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/671460/6/check/openstack-tox-pep8/71de206/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/60/671460/6/check/openstack-tox-py27/99866ee/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/60/671460/6/check/openstack-tox-py36/47c555c/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/60/671460/6/check/openstack-tox-py37/8678c85/ : SUCCESS in 5m 29s\n- openstack-tox-molecule http://logs.openstack.org/60/671460/6/check/openstack-tox-molecule/7717f32/tox/reports.html : SUCCESS in 13m 35s\n- openstack-tox-docs http://logs.openstack.org/60/671460/6/check/openstack-tox-docs/b379f40/html/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/671460/6/check/openstack-tox-lower-constraints/efe592e/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5f54c3473426ab6052a01aab7db7b1940b2665b7","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-26 06:32:49.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1be56963818965060053f9be66f55a95fd688c4e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-26 07:59:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6f4f46b1ef926bfcf00c10a11d41c89bfcacf5e9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 10:29:52.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f4394ca80cd81b03e8e78365f3cb9207d1372542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 14:08:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-containers-multinode/a3ac0d9/ : SUCCESS in 2h 19m 53s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-undercloud-containers/35e225c/ : SUCCESS in 1h 44m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/94e68cd/ : SUCCESS in 2h 25m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/334669e/ : TIMED_OUT in 3h 21m 17s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-standalone/38ea4c5/ : SUCCESS in 1h 57m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/60/671460/6/check/tripleo-ci-fedora-28-standalone/ffee569/ : SUCCESS in 1h 54m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/60/671460/6/check/tripleo-ci-centos-7-standalone-upgrade/628eb42/ : SUCCESS in 2h 16m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/671460/6/check/openstack-tox-pep8/0d57928/ : SUCCESS in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/60/671460/6/check/openstack-tox-py27/66b8a23/ : SUCCESS in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/60/671460/6/check/openstack-tox-py36/3d2defb/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/60/671460/6/check/openstack-tox-py37/d61a267/ : SUCCESS in 4m 45s\n- openstack-tox-molecule http://logs.openstack.org/60/671460/6/check/openstack-tox-molecule/4b4f243/tox/reports.html : SUCCESS in 27m 30s\n- openstack-tox-docs http://logs.openstack.org/60/671460/6/check/openstack-tox-docs/4c2eba4/html/ : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/671460/6/check/openstack-tox-lower-constraints/d369416/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":6},{"id":"baa95a6958b24b3db66ed7448c6155a60f5ebaf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 13:57:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dde50521513870f119ac2417d8292c74ecf2e828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 17:40:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"cb09e87f16a1596ecb4a69a510517b74a027b73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 17:40:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/60/671460/6/gate/tripleo-ci-centos-7-containers-multinode/2d7c57f/ : SUCCESS in 2h 26m 42s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/60/671460/6/gate/tripleo-ci-centos-7-standalone/f3f0b0d/ : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/60/671460/6/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/205c1bc/ : SUCCESS in 2h 23m 28s\n- openstack-tox-pep8 http://logs.openstack.org/60/671460/6/gate/openstack-tox-pep8/3454a80/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/60/671460/6/gate/openstack-tox-py27/7f02da5/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/60/671460/6/gate/openstack-tox-py36/58345c4/ : SUCCESS in 6m 57s\n- openstack-tox-py37 http://logs.openstack.org/60/671460/6/gate/openstack-tox-py37/02851ad/ : SUCCESS in 6m 12s\n- openstack-tox-molecule http://logs.openstack.org/60/671460/6/gate/openstack-tox-molecule/8575ec1/tox/reports.html : SUCCESS in 27m 56s\n- openstack-tox-docs http://logs.openstack.org/60/671460/6/gate/openstack-tox-docs/690b02b/html/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/671460/6/gate/openstack-tox-lower-constraints/4113bb7/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":6},{"id":"00b2f16f03b9b07d87b3d96fd1ce41abacbaccdd","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 10:34:06.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 000718fa93aced3beea549ee0c810572a1836168","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2affbb11f82afcdfcd77c282f718c92595cad291","revisions":{"9ae2a503f3aa97af85698444250a3b7229d6a130":{"kind":"REWORK","_number":1,"created":"2019-07-18 11:26:44.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/1"}}},"commit":{"parents":[{"commit":"bc8d8e6aa1bb99e038b1e77dceba157f63edd429","subject":"Merge \"Use ansible_pkg_mgr fact instead of hardcoding yum command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bc8d8e6aa1bb99e038b1e77dceba157f63edd429"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"subject":"Small optimization for validations run: ignore facts","message":"Small optimization for validations run: ignore facts\n\nMost of the validations don\u0027t rely on the facts, although they are\ngathered by default.\n\nThis patch intends to disable fact gathering for most of the\nvalidations, while keeping the two only ones using them.\n\nThis means a ~4s gain on a validation run, which is already a good\npoint.\n\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9ae2a503f3aa97af85698444250a3b7229d6a130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9ae2a503f3aa97af85698444250a3b7229d6a130"}]},"branch":"refs/heads/master"},"fd58639f44db85ea63bcb681b93452af5c21f042":{"kind":"REWORK","_number":2,"created":"2019-07-18 12:20:08.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/2"}}},"commit":{"parents":[{"commit":"bc8d8e6aa1bb99e038b1e77dceba157f63edd429","subject":"Merge \"Use ansible_pkg_mgr fact instead of hardcoding yum command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bc8d8e6aa1bb99e038b1e77dceba157f63edd429"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 12:19:59.000000000","tz":120},"subject":"Small optimization for validations run: ignore facts","message":"Small optimization for validations run: ignore facts\n\nMost of the validations don\u0027t rely on the facts, although they are\ngathered by default.\n\nThis patch intends to disable fact gathering for most of the\nvalidations, while keeping the two only ones using them.\n\nThis means a ~4s gain on a validation run, which is already a good\npoint.\n\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fd58639f44db85ea63bcb681b93452af5c21f042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fd58639f44db85ea63bcb681b93452af5c21f042"}]},"branch":"refs/heads/master"},"a973ad4da43819f701e2aaf2ca633f53eede3a37":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-22 07:48:37.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/3"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 07:48:20.000000000","tz":120},"subject":"Small optimization for validations run: ignore facts","message":"Small optimization for validations run: ignore facts\n\nMost of the validations don\u0027t rely on the facts, although they are\ngathered by default.\n\nThis patch intends to disable fact gathering for most of the\nvalidations, while keeping the two only ones using them.\n\nThis means a ~4s gain on a validation run, which is already a good\npoint.\n\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a973ad4da43819f701e2aaf2ca633f53eede3a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a973ad4da43819f701e2aaf2ca633f53eede3a37"}]},"branch":"refs/heads/master"},"056197a07cfd02000fbd49390861635ae821d966":{"kind":"REWORK","_number":4,"created":"2019-07-23 13:07:56.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/4"}}},"commit":{"parents":[{"commit":"9bbd3b26df331ad7307d8f0fe69894277eafdecc","subject":"Adding new role automation with Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 13:06:12.000000000","tz":120},"subject":"Enable fact gathering for needed validations","message":"Enable fact gathering for needed validations\n\nImplicit fact gathering is disabled since\nhttps://review.opendev.org/672278\n\nThis means, for some validations, we must explicitely enable fact\ngathering.\n\nDepends-On: https://review.opendev.org/672278\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/056197a07cfd02000fbd49390861635ae821d966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/056197a07cfd02000fbd49390861635ae821d966"}]},"branch":"refs/heads/master"},"a10f7d323e2d04f805c03a8b7e7897562ceae963":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-23 13:09:21.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/5"}}},"commit":{"parents":[{"commit":"9bbd3b26df331ad7307d8f0fe69894277eafdecc","subject":"Adding new role automation with Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 13:09:08.000000000","tz":120},"subject":"Enable fact gathering for needed validations","message":"Enable fact gathering for needed validations\n\nImplicit fact gathering is disabled since\nhttps://review.opendev.org/672278\n\nThis means, for some validations, we must explicitly enable fact\ngathering.\n\nDepends-On: https://review.opendev.org/672278\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a10f7d323e2d04f805c03a8b7e7897562ceae963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a10f7d323e2d04f805c03a8b7e7897562ceae963"}]},"branch":"refs/heads/master"},"2affbb11f82afcdfcd77c282f718c92595cad291":{"kind":"REWORK","_number":6,"created":"2019-07-23 13:58:12.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/671460/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/60/671460/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/60/671460/6"}}},"commit":{"parents":[{"commit":"9bbd3b26df331ad7307d8f0fe69894277eafdecc","subject":"Adding new role automation with Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 13:58:04.000000000","tz":120},"subject":"Enable fact gathering for needed validations","message":"Enable fact gathering for needed validations\n\nImplicit fact gathering is disabled since\nhttps://review.opendev.org/672278\n\nThis means, for some validations, we must explicitly enable fact\ngathering.\n\nDepends-On: https://review.opendev.org/672278\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2affbb11f82afcdfcd77c282f718c92595cad291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2affbb11f82afcdfcd77c282f718c92595cad291"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
