)]}'
{"id":"openstack%2Ftripleo-validations~672003","triplet_id":"openstack%2Ftripleo-validations~master~Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da","project":"openstack/tripleo-validations","branch":"master","topic":"molecule/dns","hashtags":[],"change_id":"Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da","subject":"Adds molecule testing for DNS validation","status":"MERGED","created":"2019-07-22 09:48:30.000000000","updated":"2019-08-01 11:05:34.000000000","submitted":"2019-07-23 17:42:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672003-1563903770196-e6e18664","meta_rev_id":"b89ce9e19968a4be030215bb1ca098b58e6f5500","_number":672003,"virtual_id_number":672003,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-23 17:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2019-07-22 14:18:51.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-07-23 14:20:49.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-07-22 13:22:15.000000000","_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"date":"2019-07-22 13:36:50.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-07-23 14:20:49.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 13:22:15.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-22 13:36:50.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-22 14:18:51.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-23 14:20:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 17:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f659c8bd8cb0e67a8c4cd8948b75f9f569366517","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-22 09:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afafdce96f0a1fa8863b0cd8e0cf2634023a0af9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:02:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-containers-multinode/0b41b74/ : SUCCESS in 2h 24m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-undercloud-containers/2cd5e37/ : SUCCESS in 1h 48m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/23efd57/ : SUCCESS in 2h 31m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/43c8274/ : SUCCESS in 2h 30m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-standalone/6dc866c/ : SUCCESS in 1h 51m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/03/672003/1/check/tripleo-ci-fedora-28-standalone/3a29c0c/ : SUCCESS in 1h 48m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/03/672003/1/check/tripleo-ci-centos-7-standalone-upgrade/32f2f3c/ : FAILURE in 1h 50m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/03/672003/1/check/openstack-tox-pep8/8241df6/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/03/672003/1/check/openstack-tox-py27/e174a29/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/03/672003/1/check/openstack-tox-py36/fa8f24e/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/03/672003/1/check/openstack-tox-py37/4205d75/ : SUCCESS in 5m 21s\n- openstack-tox-molecule http://logs.openstack.org/03/672003/1/check/openstack-tox-molecule/aeb9fec/tox/reports.html : SUCCESS in 13m 58s\n- openstack-tox-docs http://logs.openstack.org/03/672003/1/check/openstack-tox-docs/c33d3d8/html/ : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/672003/1/check/openstack-tox-lower-constraints/820e699/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f93e5235a2e83e9747f099169f1eb11808477c6b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 14:18:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c81d945be8a0bf3a9377880fbe8953b02fb691db","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 14:20:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46040d847711f03293f2e12fc76edf674d43faee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:21:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4baaa1c942e216b47062a73c52ae1ea7ccac47a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:42:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/672003/1/gate/tripleo-ci-centos-7-containers-multinode/9d73d52/ : SUCCESS in 2h 29m 39s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/03/672003/1/gate/tripleo-ci-centos-7-standalone/45ae0c1/ : SUCCESS in 1h 52m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/672003/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/483c74a/ : SUCCESS in 2h 25m 39s\n- openstack-tox-pep8 http://logs.openstack.org/03/672003/1/gate/openstack-tox-pep8/e775095/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/03/672003/1/gate/openstack-tox-py27/57810f7/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/03/672003/1/gate/openstack-tox-py36/684b59f/ : SUCCESS in 6m 29s\n- openstack-tox-py37 http://logs.openstack.org/03/672003/1/gate/openstack-tox-py37/343e424/ : SUCCESS in 5m 35s\n- openstack-tox-molecule http://logs.openstack.org/03/672003/1/gate/openstack-tox-molecule/01f7962/tox/reports.html : SUCCESS in 12m 34s\n- openstack-tox-docs http://logs.openstack.org/03/672003/1/gate/openstack-tox-docs/3a97440/html/ : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/672003/1/gate/openstack-tox-lower-constraints/983557b/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5215edcf327865ff19432486ac3be79c6d1e5db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:42:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c4bc30256c86c051f2d2258f949e6b059fa586a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:05:34.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 9b57fad89464e65cb26dcda7d4ea8105511615df","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"647e9afc264fab5a0672d372d7244a171104486e","revisions":{"647e9afc264fab5a0672d372d7244a171104486e":{"kind":"REWORK","_number":1,"created":"2019-07-22 09:48:30.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/03/672003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/03/672003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/03/672003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/03/672003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/03/672003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/03/672003/1"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 09:48:17.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 09:48:17.000000000","tz":120},"subject":"Adds molecule testing for DNS validation","message":"Adds molecule testing for DNS validation\n\nChange-Id: Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/647e9afc264fab5a0672d372d7244a171104486e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/647e9afc264fab5a0672d372d7244a171104486e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
