)]}'
{"id":"openstack%2Ftripleo-validations~672032","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I5dc5ad09684a90ff4f98be3ae570c85c2937211a","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"backup-jun24-stable/stein","hashtags":[],"change_id":"I5dc5ad09684a90ff4f98be3ae570c85c2937211a","subject":"Fix unittests for Stein","status":"MERGED","created":"2019-07-22 12:43:34.000000000","updated":"2019-07-23 17:00:58.000000000","submitted":"2019-07-23 17:00:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672032-1563901258071-b5d406da","meta_rev_id":"aad748011cb0f383c469ae13e5ac8ea7c7f9761d","_number":672032,"virtual_id_number":672032,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-23 17:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2019-07-23 13:22:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 14:11:27.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 12:44:04.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-23 13:22:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 14:11:27.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-23 17:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65761147a0f93bcc62962e56f67d964a5a17d65e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 12:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18e523d640f6f11c52c49e96fd8e38c7e0cd13b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:44:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672032/1/check/tripleo-ci-centos-7-containers-multinode/7245bf0/ : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/672032/1/check/tripleo-ci-centos-7-undercloud-containers/9fd7917/ : SUCCESS in 1h 45m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/672032/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/dbebd09/ : SUCCESS in 2h 41m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/672032/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5ee0737/ : SUCCESS in 2h 51m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672032/1/check/tripleo-ci-centos-7-standalone/3fe6ee3/ : SUCCESS in 2h 01m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/672032/1/check/tripleo-ci-fedora-28-standalone/50907eb/ : FAILURE in 1h 52m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/32/672032/1/check/openstack-tox-pep8/44c1953/ : FAILURE in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/32/672032/1/check/openstack-tox-py27/8f3e004/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/32/672032/1/check/openstack-tox-py35/d5fedff/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/32/672032/1/check/openstack-tox-py36/1bce64f/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/32/672032/1/check/openstack-tox-docs/98ae78c/html/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672032/1/check/openstack-tox-lower-constraints/5f248bc/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c14457c3b10103b0ac212a4d3329ecf0edb20034","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 09:46:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"73073230381f427665993a55ca542b949b17861f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:45:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672032/2/check/tripleo-ci-centos-7-containers-multinode/ee8fc73/ : SUCCESS in 2h 21m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/672032/2/check/tripleo-ci-centos-7-undercloud-containers/0062020/ : SUCCESS in 2h 00m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/672032/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/77e20e4/ : SUCCESS in 2h 35m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/672032/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/743cbca/ : SUCCESS in 2h 53m 12s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672032/2/check/tripleo-ci-centos-7-standalone/18402b1/ : SUCCESS in 2h 10m 58s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/672032/2/check/tripleo-ci-fedora-28-standalone/f83aa88/ : SUCCESS in 1h 53m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/32/672032/2/check/openstack-tox-pep8/f80f87f/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/32/672032/2/check/openstack-tox-py27/0fa5853/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/32/672032/2/check/openstack-tox-py35/e88cc5d/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/32/672032/2/check/openstack-tox-py36/5fab8f0/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/32/672032/2/check/openstack-tox-docs/722d081/html/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672032/2/check/openstack-tox-lower-constraints/5a64dab/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"a081f092a5c9dc18e5c81eae21f94450988aad5e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 13:22:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbd8769deb150132b6d132620e0d3e936a135ca9","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 14:11:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d0f914a9418d3fae65706b517d0c35037fee321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:11:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8da7057f671f43252615aabfc00c21554accf92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:00:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672032/2/gate/tripleo-ci-centos-7-containers-multinode/06e1087/ : SUCCESS in 2h 27m 49s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672032/2/gate/tripleo-ci-centos-7-standalone/f86744a/ : SUCCESS in 2h 05m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/672032/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/61d36fb/ : SUCCESS in 2h 41m 32s\n- openstack-tox-pep8 http://logs.openstack.org/32/672032/2/gate/openstack-tox-pep8/93a0d0f/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/32/672032/2/gate/openstack-tox-py27/b187b98/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/32/672032/2/gate/openstack-tox-py35/c742c9e/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/32/672032/2/gate/openstack-tox-py36/60410e6/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/32/672032/2/gate/openstack-tox-docs/9efddae/html/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672032/2/gate/openstack-tox-lower-constraints/abea491/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b5278598f5e5d05c6274394f9b8dd9454a12d6ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:00:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7aa56fa47c6f9159fa808aa34c74585196fdbdd7","revisions":{"24476bf8e5db36ab2ffddcb760520ee19028515c":{"kind":"REWORK","_number":1,"created":"2019-07-22 12:43:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/32/672032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/32/672032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/1"}}},"commit":{"parents":[{"commit":"f3fe2085e10b7b7dcb345762988013436e8a8edd","subject":"Use ansible_pkg_mgr fact instead of hardcoding yum command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f3fe2085e10b7b7dcb345762988013436e8a8edd"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 12:35:04.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 12:43:09.000000000","tz":120},"subject":"Fix unittests for Stein","message":"Fix unittests for Stein\n\nWe still were testing validations dir contents which is deprecated in\nstable/stein. This patch fixes the unittests by testing the right\nlibraries from the right place.\n\nChange-Id: I5dc5ad09684a90ff4f98be3ae570c85c2937211a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/24476bf8e5db36ab2ffddcb760520ee19028515c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/24476bf8e5db36ab2ffddcb760520ee19028515c"}]},"branch":"refs/heads/stable/stein"},"7aa56fa47c6f9159fa808aa34c74585196fdbdd7":{"kind":"REWORK","_number":2,"created":"2019-07-23 09:46:12.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/32/672032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/32/672032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/32/672032/2"}}},"commit":{"parents":[{"commit":"f3fe2085e10b7b7dcb345762988013436e8a8edd","subject":"Use ansible_pkg_mgr fact instead of hardcoding yum command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f3fe2085e10b7b7dcb345762988013436e8a8edd"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 12:35:04.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 09:46:08.000000000","tz":0},"subject":"Fix unittests for Stein","message":"Fix unittests for Stein\n\nWe still were testing validations dir contents which is deprecated in\nstable/stein. This patch fixes the unittests by testing the right\nlibraries from the right place.\n\nChange-Id: I5dc5ad09684a90ff4f98be3ae570c85c2937211a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7aa56fa47c6f9159fa808aa34c74585196fdbdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7aa56fa47c6f9159fa808aa34c74585196fdbdd7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
