)]}'
{"id":"openstack%2Ftripleo-validations~672050","triplet_id":"openstack%2Ftripleo-validations~master~I20b55dc55b833a87930a4d9efd9fe81d4759ccb5","project":"openstack/tripleo-validations","branch":"master","topic":"molecule/undercloud-disk-space","hashtags":[],"change_id":"I20b55dc55b833a87930a4d9efd9fe81d4759ccb5","subject":"Adds molecule tests for undercloud-disk-space","status":"MERGED","created":"2019-07-22 14:05:30.000000000","updated":"2019-08-01 11:06:03.000000000","submitted":"2019-07-23 15:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672050-1563895571231-cc4310e6","meta_rev_id":"f1672fc67b0813d23649aa97127b720663101800","_number":672050,"virtual_id_number":672050,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-23 15:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-23 10:26:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":2,"date":"2019-07-22 14:19:47.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-23 10:26:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"date":"2019-07-23 10:28:06.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-07-22 14:06:02.000000000","_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"date":"2019-07-22 14:06:10.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-07-23 13:06:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-23 10:28:16.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 14:06:02.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-22 14:06:10.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-22 14:19:47.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-23 10:26:19.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-07-23 10:26:49.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-23 10:28:06.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-07-23 10:28:16.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-07-23 13:06:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 15:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70ca1be7fa1683f8743917694ff1a81ae8b08717","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-22 14:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90b2f8510cc85bcdc4bf48c61cd47e993b9dd8b5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 14:19:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e83a700cc391cd08c309ebfdcc840e909815863d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-containers-multinode/af4f16e/ : SUCCESS in 2h 49m 27s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-undercloud-containers/31efbbd/ : SUCCESS in 1h 38m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/777ac16/ : SUCCESS in 2h 12m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c61807c/ : SUCCESS in 2h 57m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-standalone/490c0cf/ : SUCCESS in 1h 56m 26s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/50/672050/1/check/tripleo-ci-fedora-28-standalone/5ba22bb/ : SUCCESS in 1h 42m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/50/672050/1/check/tripleo-ci-centos-7-standalone-upgrade/6a69751/ : FAILURE in 1h 54m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/672050/1/check/openstack-tox-pep8/db5c491/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/50/672050/1/check/openstack-tox-py27/97d7370/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/50/672050/1/check/openstack-tox-py36/9b45347/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/50/672050/1/check/openstack-tox-py37/d37d8ae/ : SUCCESS in 6m 16s\n- openstack-tox-molecule http://logs.openstack.org/50/672050/1/check/openstack-tox-molecule/fa78772/tox/reports.html : SUCCESS in 13m 24s\n- openstack-tox-docs http://logs.openstack.org/50/672050/1/check/openstack-tox-docs/3545c74/html/ : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/672050/1/check/openstack-tox-lower-constraints/d7a4f05/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d79f0a5908e99e2ec31c6e7f0707ad7b0315002d","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-23 10:26:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8169bde8070a249e08ff50e2e4b3f3c5e35196c3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 13:06:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af079e7a3ab2ffc9217c94d1afa059d8374f66a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:06:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2d0b998bfd02ea29acd43483e9e62217593296d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:26:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/672050/1/gate/tripleo-ci-centos-7-containers-multinode/392ee71/ : SUCCESS in 2h 12m 49s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/50/672050/1/gate/tripleo-ci-centos-7-standalone/fe130c1/ : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/50/672050/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0374ded/ : SUCCESS in 2h 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/50/672050/1/gate/openstack-tox-pep8/82094b7/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/50/672050/1/gate/openstack-tox-py27/f7134ff/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/50/672050/1/gate/openstack-tox-py36/2554793/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/50/672050/1/gate/openstack-tox-py37/84e53a7/ : SUCCESS in 6m 17s\n- openstack-tox-molecule http://logs.openstack.org/50/672050/1/gate/openstack-tox-molecule/6dc6db6/tox/reports.html : SUCCESS in 12m 17s\n- openstack-tox-docs http://logs.openstack.org/50/672050/1/gate/openstack-tox-docs/970e7b8/html/ : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/672050/1/gate/openstack-tox-lower-constraints/887d9e3/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8ee04aaf33591178bc46bef2e001caf3def7ca70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:26:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c75a983c8db253dcb188c503508975907c6652ba","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:06:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a95ad30c28db5551c8a57d738cd328ceda8765ed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5adfef9d9a61402617591ef58315b9cd62f60907","revisions":{"5adfef9d9a61402617591ef58315b9cd62f60907":{"kind":"REWORK","_number":1,"created":"2019-07-22 14:05:30.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/50/672050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/50/672050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/50/672050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/50/672050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/50/672050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/50/672050/1"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 14:05:22.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 14:05:22.000000000","tz":120},"subject":"Adds molecule tests for undercloud-disk-space","message":"Adds molecule tests for undercloud-disk-space\n\nChange-Id: I20b55dc55b833a87930a4d9efd9fe81d4759ccb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5adfef9d9a61402617591ef58315b9cd62f60907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5adfef9d9a61402617591ef58315b9cd62f60907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
