)]}'
{"id":"openstack%2Ftripleo-validations~672052","triplet_id":"openstack%2Ftripleo-validations~master~I483d5e7b11376dc0141c506bf07ccd5876e4076e","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework","hashtags":[],"change_id":"I483d5e7b11376dc0141c506bf07ccd5876e4076e","subject":"Adding new role automation with Ansible","status":"MERGED","created":"2019-07-22 14:12:28.000000000","updated":"2019-07-23 10:42:11.000000000","submitted":"2019-07-23 10:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672052-1563878531224-4e3e88c1","meta_rev_id":"d1f7ffe16097617170842062782083737b199440","_number":672052,"virtual_id_number":672052,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-23 10:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-22 19:09:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 08:08:05.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 08:08:05.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 14:17:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-22 19:09:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 08:08:05.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-23 10:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4261b3702bc748f520bba9be8a53cd112bf5fd58","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 14:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1e747ff5a6eceb2940fb31d6d7fd15eed9c7086","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 14:16:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0d250cab2010ce06dc79938d1ef6f0c46d7a80d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 14:17:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f94baa4b975026857228777055a0e48b0e41748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:56:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-containers-multinode/40f0b68/ : SUCCESS in 2h 29m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-undercloud-containers/5fb76ed/ : SUCCESS in 1h 38m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/748078d/ : SUCCESS in 2h 24m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b26e33e/ : SUCCESS in 2h 51m 36s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-standalone/52e689f/ : SUCCESS in 1h 39m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/52/672052/3/check/tripleo-ci-fedora-28-standalone/7f834fd/ : SUCCESS in 1h 44m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/52/672052/3/check/tripleo-ci-centos-7-standalone-upgrade/5d6a80e/ : FAILURE in 2h 04m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/52/672052/3/check/openstack-tox-pep8/e4aae5a/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/52/672052/3/check/openstack-tox-py27/953c463/ : SUCCESS in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/52/672052/3/check/openstack-tox-py36/d966835/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/52/672052/3/check/openstack-tox-py37/077a639/ : SUCCESS in 5m 18s\n- openstack-tox-molecule http://logs.openstack.org/52/672052/3/check/openstack-tox-molecule/90c5f61/tox/reports.html : SUCCESS in 10m 36s\n- openstack-tox-docs http://logs.openstack.org/52/672052/3/check/openstack-tox-docs/5c5b354/html/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/672052/3/check/openstack-tox-lower-constraints/5034627/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1eec9b5ddf57eca4a1aed77afc66cb1bcf091847","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-22 19:09:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c6a4bcbcc1f7992f3bf07dc1bc064fdd8be563f1","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 08:08:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"113b587b2adbbb286d0718c16eaffe0f0220ce80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:08:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"342119ee739d3ef58680e8bc2a806af8b17a5655","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:42:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b71d462d512c37c5968b969ea5e9b2c317f51b63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:42:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/52/672052/3/gate/tripleo-ci-centos-7-containers-multinode/54f2f2c/ : SUCCESS in 2h 17m 46s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/52/672052/3/gate/tripleo-ci-centos-7-standalone/62e401e/ : SUCCESS in 1h 43m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/52/672052/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a37ab94/ : SUCCESS in 2h 09m 38s\n- openstack-tox-pep8 http://logs.openstack.org/52/672052/3/gate/openstack-tox-pep8/c2db7a3/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/52/672052/3/gate/openstack-tox-py27/1be0dfc/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/52/672052/3/gate/openstack-tox-py36/81d60bc/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/52/672052/3/gate/openstack-tox-py37/74b7eb6/ : SUCCESS in 8m 23s\n- openstack-tox-molecule http://logs.openstack.org/52/672052/3/gate/openstack-tox-molecule/86a9415/tox/reports.html : SUCCESS in 10m 35s\n- openstack-tox-docs http://logs.openstack.org/52/672052/3/gate/openstack-tox-docs/f86bec8/html/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/672052/3/gate/openstack-tox-lower-constraints/1464ca6/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9bbd3b26df331ad7307d8f0fe69894277eafdecc","revisions":{"57e4844a424a639b08ed5a2be16e8de1a0088dba":{"kind":"REWORK","_number":1,"created":"2019-07-22 14:12:28.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/52/672052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/52/672052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/1"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:04:23.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:11:26.000000000","tz":120},"subject":"Adding new role automation with Ansible","message":"Adding new role automation with Ansible\n\nThis patch adds a basic skeleton to create a new roles including\nmolecule structure.\n\nChange-Id: I483d5e7b11376dc0141c506bf07ccd5876e4076e\nPatch-Inspired-From: tripleo-ansible-Team++\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/57e4844a424a639b08ed5a2be16e8de1a0088dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/57e4844a424a639b08ed5a2be16e8de1a0088dba"}]},"branch":"refs/heads/master"},"186b629fd7cfb05d5a5e58bbfb9e5a3a7c4b81a2":{"kind":"REWORK","_number":2,"created":"2019-07-22 14:16:41.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/52/672052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/52/672052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/2"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:04:23.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:16:31.000000000","tz":120},"subject":"Adding new role automation with Ansible","message":"Adding new role automation with Ansible\n\nThis patch adds a basic skeleton to create a new roles including\nmolecule structure.\n\nChange-Id: I483d5e7b11376dc0141c506bf07ccd5876e4076e\nPatch-Inspired-From: tripleo-ansible-Team++\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/186b629fd7cfb05d5a5e58bbfb9e5a3a7c4b81a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/186b629fd7cfb05d5a5e58bbfb9e5a3a7c4b81a2"}]},"branch":"refs/heads/master"},"9bbd3b26df331ad7307d8f0fe69894277eafdecc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-22 14:17:30.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/52/672052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/52/672052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/52/672052/3"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:04:23.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:17:27.000000000","tz":0},"subject":"Adding new role automation with Ansible","message":"Adding new role automation with Ansible\n\nThis patch adds a basic skeleton to create a new roles including\nmolecule structure.\n\nChange-Id: I483d5e7b11376dc0141c506bf07ccd5876e4076e\nPatch-Inspired-By: tripleo-ansible-Team++\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
