)]}'
{"id":"openstack%2Ftripleo-validations~672205","triplet_id":"openstack%2Ftripleo-validations~master~Ib8ae41451f245d4e2d40864cc36515a07a1878bf","project":"openstack/tripleo-validations","branch":"master","topic":"molecule/ctlplane-ip-range","hashtags":[],"change_id":"Ib8ae41451f245d4e2d40864cc36515a07a1878bf","subject":"Adds molecule tests for ctlplane-ip-range validation","status":"MERGED","created":"2019-07-23 06:03:13.000000000","updated":"2019-08-01 11:07:25.000000000","submitted":"2019-07-23 18:12:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672205-1563905572839-a51fe37c","meta_rev_id":"6e7cb70890d0835b97983cdc497ed7c74c9044b9","_number":672205,"virtual_id_number":672205,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-07-23 18:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-23 12:11:42.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2019-07-23 14:18:05.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-07-23 15:28:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 06:03:36.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-23 06:03:40.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-23 14:18:05.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-23 15:28:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 18:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc37d266dee697917122d0b6e214d27f8fc84567","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 06:03:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e2373044165ec3b086d0df04e9979a2b61f78b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 07:45:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"351d1eed19bb33f183940745f8c3c5be5df26eda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:19:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-containers-multinode/ad38b93/ : FAILURE in 1h 56m 27s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-undercloud-containers/f5c45d6/ : SUCCESS in 1h 33m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/40e0fb0/ : SUCCESS in 2h 29m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c2d6241/ : SUCCESS in 2h 30m 02s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-standalone/05b5051/ : SUCCESS in 1h 45m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/672205/2/check/tripleo-ci-fedora-28-standalone/e8e2880/ : SUCCESS in 1h 37m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-standalone-upgrade/e399545/ : FAILURE in 2h 00m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/05/672205/2/check/openstack-tox-pep8/0cfa2bd/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/05/672205/2/check/openstack-tox-py27/9de4cf3/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/05/672205/2/check/openstack-tox-py36/066d775/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/05/672205/2/check/openstack-tox-py37/66a3ba5/ : SUCCESS in 4m 42s\n- openstack-tox-molecule http://logs.openstack.org/05/672205/2/check/openstack-tox-molecule/63b271f/tox/reports.html : SUCCESS in 14m 39s\n- openstack-tox-docs http://logs.openstack.org/05/672205/2/check/openstack-tox-docs/42074ac/html/ : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/672205/2/check/openstack-tox-lower-constraints/387c2af/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d10ecbc81d1647e9bd79b4555aa3dffa6ad25592","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 12:11:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a489081f534963df8daeee0a0a72a352c4d58226","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 14:18:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"624ec75b008c02f59fb3813b03129eeb5b3c28a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:49:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-containers-multinode/6797803/ : SUCCESS in 2h 16m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-undercloud-containers/3f0eeeb/ : SUCCESS in 1h 54m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/06ea909/ : SUCCESS in 2h 22m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d04bbc4/ : SUCCESS in 2h 27m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-standalone/a9dccce/ : SUCCESS in 1h 44m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/672205/2/check/tripleo-ci-fedora-28-standalone/0e3162e/ : SUCCESS in 1h 44m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/672205/2/check/tripleo-ci-centos-7-standalone-upgrade/d1b61c8/ : FAILURE in 2h 00m 19s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/05/672205/2/check/openstack-tox-pep8/b6f0430/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/05/672205/2/check/openstack-tox-py27/8a330b2/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/05/672205/2/check/openstack-tox-py36/683b4b2/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/05/672205/2/check/openstack-tox-py37/a7ce795/ : SUCCESS in 4m 53s\n- openstack-tox-molecule http://logs.openstack.org/05/672205/2/check/openstack-tox-molecule/9e3fba9/tox/reports.html : SUCCESS in 14m 13s\n- openstack-tox-docs http://logs.openstack.org/05/672205/2/check/openstack-tox-docs/7822dc0/html/ : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/672205/2/check/openstack-tox-lower-constraints/9d542c8/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1522f2e892fbab74a98cfc25b1a4bc7220e05dcc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 15:28:43.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56ca4fdf57dba6e19f84b397ef8f27f67b598ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:29:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8634f94342f682ddba1c1041fd321d08a3946a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 18:12:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/672205/2/gate/tripleo-ci-centos-7-containers-multinode/ec45dc4/ : SUCCESS in 2h 31m 20s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/672205/2/gate/tripleo-ci-centos-7-standalone/5233f00/ : SUCCESS in 1h 54m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/672205/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/377e21f/ : SUCCESS in 2h 22m 50s\n- openstack-tox-pep8 http://logs.openstack.org/05/672205/2/gate/openstack-tox-pep8/0ceec13/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/05/672205/2/gate/openstack-tox-py27/b54230f/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/05/672205/2/gate/openstack-tox-py36/074b0fe/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/05/672205/2/gate/openstack-tox-py37/0a48a57/ : SUCCESS in 5m 05s\n- openstack-tox-molecule http://logs.openstack.org/05/672205/2/gate/openstack-tox-molecule/a3b8e94/tox/reports.html : SUCCESS in 24m 07s\n- openstack-tox-docs http://logs.openstack.org/05/672205/2/gate/openstack-tox-docs/0b8fa0b/html/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/672205/2/gate/openstack-tox-lower-constraints/99aa6ac/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3faf379646120c0a49a9716aa0ce627ecd3e233c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 18:12:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b84bff4d3d1d6acf265908a074b8c24f5af1dd7","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:07:25.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d3cd7c726596dbcfc99c4927e29dd3e44b74084c","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49aa25278db5031d0543169ad2e12676ca89fc7d","revisions":{"1c2275822b88e2684bf86460c76ff36ee076dca0":{"kind":"REWORK","_number":1,"created":"2019-07-23 06:03:13.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/05/672205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/05/672205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/1"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1c2275822b88e2684bf86460c76ff36ee076dca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1c2275822b88e2684bf86460c76ff36ee076dca0"}]},"branch":"refs/heads/master"},"49aa25278db5031d0543169ad2e12676ca89fc7d":{"kind":"REWORK","_number":2,"created":"2019-07-23 07:45:01.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/05/672205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/05/672205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/05/672205/2"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 07:44:54.000000000","tz":120},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49aa25278db5031d0543169ad2e12676ca89fc7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49aa25278db5031d0543169ad2e12676ca89fc7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
