)]}'
{"id":"openstack%2Ftripleo-validations~672220","triplet_id":"openstack%2Ftripleo-validations~master~I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework-infra","hashtags":[],"change_id":"I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3","subject":"Add autodocumentation for roles and library","status":"MERGED","created":"2019-07-23 08:16:15.000000000","updated":"2019-08-05 09:40:19.000000000","submitted":"2019-08-05 09:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672220-1564998019379-520616fe","meta_rev_id":"b1501f50fdb3d28896726aac6eb07b132307df7f","_number":672220,"virtual_id_number":672220,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-05 09:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-08-02 17:59:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-04 10:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 07:10:55.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 08:55:05.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-23 13:51:32.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-07-26 14:20:20.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-02 17:59:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-04 10:19:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-05 07:10:55.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-05 09:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48641d13149085125941eee0a67b4f6314efcfe2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 08:16:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a13387ef13211bb2d38c0c400d76061d26d94e1","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 08:54:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"884c32f85a93e32eed80a378b45b0636ea7e7e82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:27:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-containers-multinode/ed38abe/ : FAILURE in 28m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-undercloud-containers/c8ebd2e/ : FAILURE in 24m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/88e2550/ : FAILURE in 32m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c038649/ : FAILURE in 26m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-standalone/d659692/ : FAILURE in 24m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/2/check/tripleo-ci-fedora-28-standalone/5f1692c/ : FAILURE in 32m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/2/check/tripleo-ci-centos-7-standalone-upgrade/9144a06/ : FAILURE in 1h 26m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/2/check/openstack-tox-pep8/53e0076/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/2/check/openstack-tox-py27/0bd966c/ : SUCCESS in 6m 31s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/2/check/openstack-tox-py36/7c525aa/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/2/check/openstack-tox-py37/a306520/ : SUCCESS in 4m 57s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/2/check/openstack-tox-molecule/118c0f7/tox/reports.html : SUCCESS in 10m 13s\n- openstack-tox-docs http://logs.openstack.org/20/672220/2/check/openstack-tox-docs/b3288da/html/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/2/check/openstack-tox-lower-constraints/096dd3f/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"42a28460242b11483786c2e9d9e48a4e23494e76","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-07-23 13:51:32.000000000","message":"Patch Set 2:\n\nneed to remove the angle bracket here https://opendev.org/openstack/tripleo-validations/src/commit/9bbd3b26df331ad7307d8f0fe69894277eafdecc/library/check_ironic_boot_config.py#L27 \nor else it renders funny\n\ni can do it in a separate patch or in this patch, up to you","accounts_in_message":[],"_revision_number":2},{"id":"35341d6d88776c258d5bfec9afbad0b319e3c14c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 14:08:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4cb800d3d9729d818ca14028400f6c153e9df6ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:35:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-containers-multinode/7a61a53/ : FAILURE in 29m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-undercloud-containers/63bee6d/ : FAILURE in 26m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8243b51/ : FAILURE in 34m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/48d2d5d/ : FAILURE in 25m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-standalone/e1015f5/ : FAILURE in 25m 04s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/3/check/tripleo-ci-fedora-28-standalone/5ab939a/ : FAILURE in 26m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/3/check/tripleo-ci-centos-7-standalone-upgrade/df5135e/ : FAILURE in 1h 20m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/3/check/openstack-tox-pep8/07d4b55/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/3/check/openstack-tox-py27/9e4a33c/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/3/check/openstack-tox-py36/c1bca10/ : SUCCESS in 4m 50s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/3/check/openstack-tox-py37/e0c9c41/ : SUCCESS in 5m 52s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/3/check/openstack-tox-molecule/fd4cb95/tox/reports.html : SUCCESS in 10m 57s\n- openstack-tox-docs http://logs.openstack.org/20/672220/3/check/openstack-tox-docs/f4807aa/html/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/3/check/openstack-tox-lower-constraints/1f3413d/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"1b090b6313508c367d479656b5a422e781fd122c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-24 07:52:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f461deaf1724e4024b782c63cf7856c9512c18ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:18:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-containers-multinode/6228658/ : FAILURE in 32m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-undercloud-containers/3dd7349/ : FAILURE in 22m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/baef9b6/ : FAILURE in 27m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/052355b/ : FAILURE in 25m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-standalone/2f884ea/ : FAILURE in 24m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/4/check/tripleo-ci-fedora-28-standalone/fc1183e/ : FAILURE in 26m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-standalone-upgrade/2a70f2c/ : FAILURE in 1h 21m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/4/check/openstack-tox-pep8/55c0c34/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/4/check/openstack-tox-py27/90d69c4/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/4/check/openstack-tox-py36/1fb7c3b/ : SUCCESS in 5m 57s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/4/check/openstack-tox-py37/42906b5/ : SUCCESS in 4m 52s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/4/check/openstack-tox-molecule/96cedee/tox/reports.html : SUCCESS in 24m 00s\n- openstack-tox-docs http://logs.openstack.org/20/672220/4/check/openstack-tox-docs/34f06af/html/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/4/check/openstack-tox-lower-constraints/a67c240/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b65029094245fd04fdbf89a2f28dd3948c7dc3e6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-25 06:17:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1d33f4f54caf08f1e5389147e2facff2335651a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 08:25:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-containers-multinode/eda6e6e/ : FAILURE in 26m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-undercloud-containers/b394eb6/ : FAILURE in 24m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7603588/ : FAILURE in 25m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c6b13ca/ : FAILURE in 28m 46s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-standalone/19bef68/ : FAILURE in 26m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/4/check/tripleo-ci-fedora-28-standalone/d6457a1/ : FAILURE in 31m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/4/check/tripleo-ci-centos-7-standalone-upgrade/f150043/ : FAILURE in 1h 24m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/4/check/openstack-tox-pep8/3dcb8e9/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/4/check/openstack-tox-py27/8cb27db/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/4/check/openstack-tox-py36/76bc453/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/4/check/openstack-tox-py37/763edf9/ : SUCCESS in 4m 26s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/4/check/openstack-tox-molecule/7e7df51/tox/reports.html : FAILURE in 9m 22s\n- openstack-tox-docs http://logs.openstack.org/20/672220/4/check/openstack-tox-docs/1922bc1/html/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/4/check/openstack-tox-lower-constraints/5f233e1/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"311ae9f00a5ac9faa14a814d64789bbfd3dbb5e2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-26 12:21:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"21da5639534596cdc3733769383a62d67c5a40cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:06:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-containers-multinode/41ce6a6/ : FAILURE in 30m 50s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-undercloud-containers/57fb233/ : FAILURE in 26m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4c2ed93/ : FAILURE in 29m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ffc4d77/ : FAILURE in 30m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-standalone/6081846/ : FAILURE in 28m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/5/check/tripleo-ci-fedora-28-standalone/77c72b9/ : FAILURE in 27m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/5/check/tripleo-ci-centos-7-standalone-upgrade/ccc4cc7/ : FAILURE in 1h 21m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/5/check/openstack-tox-pep8/80eca30/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/5/check/openstack-tox-py27/994d96e/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/5/check/openstack-tox-py36/aeceeb9/ : SUCCESS in 5m 20s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/5/check/openstack-tox-py37/ce6b2bb/ : SUCCESS in 5m 42s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/5/check/openstack-tox-molecule/ffbc28c/tox/reports.html : SUCCESS in 26m 55s\n- openstack-tox-docs http://logs.openstack.org/20/672220/5/check/openstack-tox-docs/8d48313/html/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/5/check/openstack-tox-lower-constraints/b6ef258/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"cf444e0026dc06c375fe4274063050fe20db4a07","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-28 14:09:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a725dfda6e2f198e3a4b05af5aa5c32ab02e0204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 15:43:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-containers-multinode/809298c/ : FAILURE in 23m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-undercloud-containers/97d2892/ : FAILURE in 21m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1e79418/ : FAILURE in 25m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4586fce/ : FAILURE in 24m 21s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-standalone/ae5876a/ : FAILURE in 19m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/6/check/tripleo-ci-fedora-28-standalone/96cdcd9/ : FAILURE in 21m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/6/check/tripleo-ci-centos-7-standalone-upgrade/ebbc456/ : FAILURE in 1h 21m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/6/check/openstack-tox-pep8/af6be04/ : FAILURE in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/6/check/openstack-tox-py27/898b01b/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/6/check/openstack-tox-py36/3f8dba2/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/6/check/openstack-tox-py37/1836026/ : SUCCESS in 4m 19s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/6/check/openstack-tox-molecule/227e560/tox/reports.html : SUCCESS in 24m 59s\n- openstack-tox-docs http://logs.openstack.org/20/672220/6/check/openstack-tox-docs/438f144/html/ : SUCCESS in 3m 57s\n- openstack-tox-linters http://logs.openstack.org/20/672220/6/check/openstack-tox-linters/1f36b35/ : FAILURE in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/6/check/openstack-tox-lower-constraints/c1845fe/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":6},{"id":"6d0e04f809d1618922c08bd5e44bde55a406ca66","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 10:24:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b307788cc1c38a47ddf8d3bdca996163c3e43f28","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:18:54.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"a39ecf8f561df2f702f143c3ce2307d2900c874e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 11:19:08.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":8},{"id":"d641b9de867b2a6f30a319476f357537da112430","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:21:39.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"a9492daaea4d1542a8d2c80eb76a67415f801995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 11:21:58.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":9},{"id":"861fdd6ef7081b03bb2c5eaf0f9dc647a85db2f6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:23:16.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"4e2c613ebeb9660e0164cfb649a08be8768a00f4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:36:54.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"2bf3cfb48ae9d91e8034c255cd08b2c6f727f601","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 12:27:39.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"08079e72b1e3cfc9f5de88d01dffb003e4e0bc5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 13:45:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-containers-multinode/f2870f8/ : FAILURE in 30m 21s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-undercloud-containers/b73f7a2/ : FAILURE in 24m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a30edfb/ : FAILURE in 28m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7351910/ : FAILURE in 28m 34s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-standalone/d29ab1a/ : FAILURE in 22m 12s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/12/check/tripleo-ci-fedora-28-standalone/11c1bcd/ : FAILURE in 27m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/12/check/tripleo-ci-centos-7-standalone-upgrade/9b9fe24/ : FAILURE in 1h 16m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/12/check/openstack-tox-pep8/88c00e1/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/12/check/openstack-tox-py27/e530ec2/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/12/check/openstack-tox-py36/2eb61de/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/12/check/openstack-tox-py37/2a4ff48/ : SUCCESS in 4m 52s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/12/check/openstack-tox-molecule/6d17ec3/tox/reports.html : SUCCESS in 25m 35s\n- openstack-tox-docs http://logs.openstack.org/20/672220/12/check/openstack-tox-docs/7401818/html/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/12/check/openstack-tox-lower-constraints/6d1a897/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":12},{"id":"67e693457f184074ccec2c078e2ede7990ab97c4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 09:47:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"05972edc40f257c3b24e8ed5dcd2cecfb8161a80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 12:00:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-containers-multinode/d8c3be8/ : FAILURE in 28m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-undercloud-containers/755eca3/ : FAILURE in 22m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c3aaa48/ : FAILURE in 33m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/06a26db/ : FAILURE in 33m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-standalone/5230f59/ : FAILURE in 25m 55s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/13/check/tripleo-ci-fedora-28-standalone/003b5fe/ : FAILURE in 26m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/13/check/tripleo-ci-centos-7-standalone-upgrade/3c61702/ : FAILURE in 50m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/13/check/openstack-tox-pep8/efad2a8/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/13/check/openstack-tox-py27/3150644/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/13/check/openstack-tox-py36/0d0cdf7/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/13/check/openstack-tox-py37/f3eed94/ : SUCCESS in 5m 26s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/13/check/openstack-tox-molecule/75400de/tox/reports.html : SUCCESS in 25m 11s\n- openstack-tox-docs http://logs.openstack.org/20/672220/13/check/openstack-tox-docs/8cb1e2e/html/ : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/13/check/openstack-tox-lower-constraints/fb1e1b4/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":13},{"id":"81064859545d7efe840ec678822eb52ff7868cb4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 12:15:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"794f2db5c76ba2219b05685624c4d1e4d1c79602","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 14:01:19.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"c3564bd89c8ba134755ba41389333504066865b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:36:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-containers-multinode/35370ae/ : FAILURE in 29m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-undercloud-containers/44f0bf7/ : FAILURE in 27m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/71a44fa/ : FAILURE in 27m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c97b461/ : FAILURE in 29m 39s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-standalone/1c45616/ : FAILURE in 25m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/15/check/tripleo-ci-fedora-28-standalone/625fb35/ : FAILURE in 26m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/15/check/tripleo-ci-centos-7-standalone-upgrade/6b9b3bf/ : FAILURE in 1h 26m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/15/check/openstack-tox-pep8/28169e7/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/15/check/openstack-tox-py27/63bbc9f/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/15/check/openstack-tox-py36/2eee50b/ : SUCCESS in 5m 59s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/15/check/openstack-tox-py37/972d8a5/ : SUCCESS in 3m 54s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/15/check/openstack-tox-molecule/c61578e/tox/reports.html : SUCCESS in 27m 32s\n- openstack-tox-docs http://logs.openstack.org/20/672220/15/check/openstack-tox-docs/184a8e8/html/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/15/check/openstack-tox-lower-constraints/1a7e708/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":15},{"id":"5a23dde1b7f5a3c6cd60927547ef2f06686856a6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 14:57:37.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3616e6357111160fb1e503e5575550d8da58864f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:33:02.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-containers-multinode/558ddbc/ : FAILURE in 28m 12s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-undercloud-containers/31df5ec/ : FAILURE in 33m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/036eb97/ : FAILURE in 29m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a31cefc/ : FAILURE in 26m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-standalone/6e7a734/ : FAILURE in 31m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/20/672220/16/check/tripleo-ci-fedora-28-standalone/58ec30b/ : FAILURE in 27m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/20/672220/16/check/tripleo-ci-centos-7-standalone-upgrade/60ff7ad/ : FAILURE in 1h 30m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/20/672220/16/check/openstack-tox-pep8/400cb54/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/20/672220/16/check/openstack-tox-py27/b895981/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/20/672220/16/check/openstack-tox-py36/8b395d0/ : SUCCESS in 6m 48s\n- openstack-tox-py37 http://logs.openstack.org/20/672220/16/check/openstack-tox-py37/4b7263e/ : SUCCESS in 4m 13s\n- openstack-tox-molecule http://logs.openstack.org/20/672220/16/check/openstack-tox-molecule/25e3a15/tox/reports.html : SUCCESS in 27m 27s\n- openstack-tox-docs http://logs.openstack.org/20/672220/16/check/openstack-tox-docs/af6e36a/html/ : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/672220/16/check/openstack-tox-lower-constraints/8177e12/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":16},{"id":"b050a1cb10725f5d124dcf0891330ae06d4fd2ba","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 10:35:06.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"056a8c455312aed15cd423a7ab08d92ac66237af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 13:21:17.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-containers-multinode/f28cdb2/ : SUCCESS in 2h 21m 51s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-undercloud-containers/07fd53f/ : SUCCESS in 1h 45m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f229b45/ : SUCCESS in 2h 25m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/723de52/ : SUCCESS in 2h 40m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-standalone/7b6e36a/ : SUCCESS in 1h 51m 00s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/20/672220/17/check/tripleo-ci-fedora-28-standalone/417e416/ : FAILURE in 1h 01m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/20/672220/17/check/tripleo-ci-centos-7-standalone-upgrade/b47c5bc/ : FAILURE in 1h 49m 44s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/20/672220/17/check/openstack-tox-pep8/7591e82/ : SUCCESS in 4m 05s\n- openstack-tox-py27 https://logs.opendev.org/20/672220/17/check/openstack-tox-py27/38425ac/ : SUCCESS in 3m 38s\n- openstack-tox-py36 https://logs.opendev.org/20/672220/17/check/openstack-tox-py36/642f978/ : SUCCESS in 3m 59s\n- openstack-tox-py37 https://logs.opendev.org/20/672220/17/check/openstack-tox-py37/8db4690/ : SUCCESS in 4m 28s\n- openstack-tox-molecule https://logs.opendev.org/20/672220/17/check/openstack-tox-molecule/62b5e91/tox/reports.html : SUCCESS in 26m 46s\n- openstack-tox-docs https://logs.opendev.org/20/672220/17/check/openstack-tox-docs/07d79d4/html/ : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/672220/17/check/openstack-tox-lower-constraints/08360fb/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":17},{"id":"4302cf2f9a7f71026b988c0278eb86f1a440605d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 13:31:18.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"37cadf366783abe02d01063728e06ea8edab3cd8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 14:16:02.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"1c208b77d16dd51353e91bcb6a8e945f779abd9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 17:23:27.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-containers-multinode/ee62ac1/ : SUCCESS in 2h 25m 42s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-undercloud-containers/8e84108/ : SUCCESS in 1h 41m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8a5048e/ : SUCCESS in 2h 35m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ca9dc1c/ : SUCCESS in 2h 30m 46s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-standalone/6287874/ : SUCCESS in 2h 04m 40s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/20/672220/18/check/tripleo-ci-fedora-28-standalone/3010fe6/ : FAILURE in 1h 05m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/20/672220/18/check/tripleo-ci-centos-7-standalone-upgrade/bac4da5/ : FAILURE in 1h 48m 24s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/20/672220/18/check/openstack-tox-pep8/eaa5689/ : SUCCESS in 4m 57s\n- openstack-tox-py27 https://logs.opendev.org/20/672220/18/check/openstack-tox-py27/d600870/ : SUCCESS in 3m 28s\n- openstack-tox-py36 https://logs.opendev.org/20/672220/18/check/openstack-tox-py36/e6781e9/ : SUCCESS in 3m 59s\n- openstack-tox-py37 https://logs.opendev.org/20/672220/18/check/openstack-tox-py37/5284edc/ : SUCCESS in 5m 01s\n- openstack-tox-molecule https://logs.opendev.org/20/672220/18/check/openstack-tox-molecule/4b777b9/tox/reports.html : SUCCESS in 26m 37s\n- openstack-tox-docs https://logs.opendev.org/20/672220/18/check/openstack-tox-docs/94b16e4/html/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/672220/18/check/openstack-tox-lower-constraints/b613ff2/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":18},{"id":"2ff646da3ea68950c036d13c80a95714524a69d5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 17:59:35.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"bf1c52b343e964bf09b691104a654bb4465a7371","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-04 10:19:46.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"62710633b27eefdba6a59eb62d6a8be74afef233","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-05 07:10:55.000000000","message":"Patch Set 18: Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"7151cebd1e3aa584a71716e9d9b8d57e61bb2b0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 07:11:08.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"0ff18f8bc87f8646cc1c575d360ceffbb2e6c8ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:40:19.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/20/672220/18/gate/tripleo-ci-centos-7-containers-multinode/6e7943e/ : SUCCESS in 2h 23m 19s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/20/672220/18/gate/tripleo-ci-centos-7-standalone/38f30c4/ : SUCCESS in 2h 14m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/20/672220/18/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1c093c6/ : SUCCESS in 2h 18m 22s\n- openstack-tox-pep8 https://logs.opendev.org/20/672220/18/gate/openstack-tox-pep8/f87e254/ : SUCCESS in 6m 24s\n- openstack-tox-py27 https://logs.opendev.org/20/672220/18/gate/openstack-tox-py27/e0e7a97/ : SUCCESS in 4m 38s\n- openstack-tox-py36 https://logs.opendev.org/20/672220/18/gate/openstack-tox-py36/b74d813/ : SUCCESS in 3m 48s\n- openstack-tox-py37 https://logs.opendev.org/20/672220/18/gate/openstack-tox-py37/2633aa8/ : SUCCESS in 5m 20s\n- openstack-tox-molecule https://logs.opendev.org/20/672220/18/gate/openstack-tox-molecule/8d6bfb1/tox/reports.html : SUCCESS in 27m 10s\n- openstack-tox-docs https://logs.opendev.org/20/672220/18/gate/openstack-tox-docs/ef13e3f/html/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/672220/18/gate/openstack-tox-lower-constraints/3832ff8/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":18},{"id":"22aacc6c0b7f8df9694188a97389adc209ddf8d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:40:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e","revisions":{"38980d0cd225e07674bbfe4a201052264a6924ae":{"kind":"REWORK","_number":1,"created":"2019-07-23 08:16:15.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/1"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:16:00.000000000","tz":120},"subject":"[WIP] Add autodocumentation for roles and library","message":"[WIP] Add autodocumentation for roles and library\n\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/38980d0cd225e07674bbfe4a201052264a6924ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/38980d0cd225e07674bbfe4a201052264a6924ae"}]},"branch":"refs/heads/master"},"fff54118aaa768dc3f404f22a4f6c978cf5b7409":{"kind":"REWORK","_number":2,"created":"2019-07-23 08:54:41.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/2"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:49:35.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fff54118aaa768dc3f404f22a4f6c978cf5b7409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fff54118aaa768dc3f404f22a4f6c978cf5b7409"}]},"branch":"refs/heads/master"},"270cd8851bb7d379330e951d7c434bb643c61045":{"kind":"REWORK","_number":3,"created":"2019-07-23 14:08:30.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/3"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 14:08:04.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/270cd8851bb7d379330e951d7c434bb643c61045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/270cd8851bb7d379330e951d7c434bb643c61045"}]},"branch":"refs/heads/master"},"5b810ed314971d666179d28d91b139092bc68316":{"kind":"REWORK","_number":4,"created":"2019-07-24 07:52:04.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/4"}}},"commit":{"parents":[{"commit":"680512410fdd93d2a3f2b23e570c90fcad7ff8ea","subject":"Merge \"Adds molecule tests for ctlplane-ip-range validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/680512410fdd93d2a3f2b23e570c90fcad7ff8ea"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-24 07:51:17.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5b810ed314971d666179d28d91b139092bc68316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5b810ed314971d666179d28d91b139092bc68316"}]},"branch":"refs/heads/master"},"c058dd9d735a9729087a7f27ed177eacb751a9b6":{"kind":"REWORK","_number":5,"created":"2019-07-26 12:21:52.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/5"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 12:20:07.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c058dd9d735a9729087a7f27ed177eacb751a9b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c058dd9d735a9729087a7f27ed177eacb751a9b6"}]},"branch":"refs/heads/master"},"a365ca81f144b0bed67fb53d641a18ac444931a9":{"kind":"REWORK","_number":6,"created":"2019-07-28 14:09:44.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/6"}}},"commit":{"parents":[{"commit":"6906ab9acda2ed525b3d59e4fde913cb9d6b7ed8","subject":"Linting hardening with pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6906ab9acda2ed525b3d59e4fde913cb9d6b7ed8"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-28 14:09:26.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a365ca81f144b0bed67fb53d641a18ac444931a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a365ca81f144b0bed67fb53d641a18ac444931a9"}]},"branch":"refs/heads/master"},"0851d91af936fd3b56459dab1293afbd8547379f":{"kind":"REWORK","_number":7,"created":"2019-07-29 10:24:57.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/7"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 10:21:14.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0851d91af936fd3b56459dab1293afbd8547379f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0851d91af936fd3b56459dab1293afbd8547379f"}]},"branch":"refs/heads/master"},"323e2f77879f11b88ac5e661da178778b845f743":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-29 11:18:54.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/8"}}},"commit":{"parents":[{"commit":"34e18e4bb54b1b9fd97a8ade19e63600ad3c51c0","subject":"[WIP] Convert tox to native zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/34e18e4bb54b1b9fd97a8ade19e63600ad3c51c0"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:16:57.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/323e2f77879f11b88ac5e661da178778b845f743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/323e2f77879f11b88ac5e661da178778b845f743"}]},"branch":"refs/heads/master"},"c4a32be1e98eea90e866dd56443da626ec439591":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-07-29 11:21:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/9"}}},"commit":{"parents":[{"commit":"d669854f06818c319f1f9310bf6b6c6152230575","subject":"[WIP] Convert tox to native zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d669854f06818c319f1f9310bf6b6c6152230575"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:21:24.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c4a32be1e98eea90e866dd56443da626ec439591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c4a32be1e98eea90e866dd56443da626ec439591"}]},"branch":"refs/heads/master"},"4b8ee8f5baab2f79c51f2e4fd5aafd7c22f23fde":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-07-29 11:23:16.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/10"}}},"commit":{"parents":[{"commit":"0ddb568b449added717989293ceff6428f01cf57","subject":"[WIP] Convert tox to native zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0ddb568b449added717989293ceff6428f01cf57"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:23:07.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4b8ee8f5baab2f79c51f2e4fd5aafd7c22f23fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4b8ee8f5baab2f79c51f2e4fd5aafd7c22f23fde"}]},"branch":"refs/heads/master"},"d8850e0c4aa84f70f04cf729d16685564012c305":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-07-29 11:36:54.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/11"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:35:35.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8850e0c4aa84f70f04cf729d16685564012c305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8850e0c4aa84f70f04cf729d16685564012c305"}]},"branch":"refs/heads/master"},"7abcd6e08e9b171fb9eded6433703a1423d1f500":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-07-29 12:27:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/12"}}},"commit":{"parents":[{"commit":"90f477c42796d68481f6a442f4e1cc5ca1c5906b","subject":"Corrected how hiera is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/90f477c42796d68481f6a442f4e1cc5ca1c5906b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 12:24:39.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7abcd6e08e9b171fb9eded6433703a1423d1f500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7abcd6e08e9b171fb9eded6433703a1423d1f500"}]},"branch":"refs/heads/master"},"876da38327115135d2deccf965e1e08d29f124f7":{"kind":"REWORK","_number":13,"created":"2019-07-30 09:47:38.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/13"}}},"commit":{"parents":[{"commit":"90f477c42796d68481f6a442f4e1cc5ca1c5906b","subject":"Corrected how hiera is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/90f477c42796d68481f6a442f4e1cc5ca1c5906b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:33:45.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/876da38327115135d2deccf965e1e08d29f124f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/876da38327115135d2deccf965e1e08d29f124f7"}]},"branch":"refs/heads/master"},"9c203f81fe270ea9a9165940268f4320a130bf3c":{"kind":"REWORK","_number":14,"created":"2019-07-30 12:15:21.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/14"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 11:53:54.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9c203f81fe270ea9a9165940268f4320a130bf3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9c203f81fe270ea9a9165940268f4320a130bf3c"}]},"branch":"refs/heads/master"},"cf2cc3c42fc963c8cf9d5caeeace9fc3ad1cf0cd":{"kind":"NO_CHANGE","_number":15,"created":"2019-07-30 14:01:19.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/15"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 13:52:34.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cf2cc3c42fc963c8cf9d5caeeace9fc3ad1cf0cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cf2cc3c42fc963c8cf9d5caeeace9fc3ad1cf0cd"}]},"branch":"refs/heads/master"},"22a78e6afbe84e30d0d4ab9e519edf12b55a5998":{"kind":"REWORK","_number":16,"created":"2019-07-31 14:57:37.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/16"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 14:35:25.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/22a78e6afbe84e30d0d4ab9e519edf12b55a5998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/22a78e6afbe84e30d0d4ab9e519edf12b55a5998"}]},"branch":"refs/heads/master"},"5f53b804f5752904e96cc4ad39613a119277f8cc":{"kind":"REWORK","_number":17,"created":"2019-08-02 10:35:06.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/17"}}},"commit":{"parents":[{"commit":"0204d7665ccec58d17feae46d9f288a638fad804","subject":"Merge \"Enable fact gathering for needed validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0204d7665ccec58d17feae46d9f288a638fad804"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-02 10:34:51.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5f53b804f5752904e96cc4ad39613a119277f8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5f53b804f5752904e96cc4ad39613a119277f8cc"}]},"branch":"refs/heads/master"},"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e":{"kind":"REWORK","_number":18,"created":"2019-08-02 14:16:02.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/20/672220/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/672220/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/672220/18"}}},"commit":{"parents":[{"commit":"0204d7665ccec58d17feae46d9f288a638fad804","subject":"Merge \"Enable fact gathering for needed validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0204d7665ccec58d17feae46d9f288a638fad804"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-23 08:14:29.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-02 14:10:47.000000000","tz":120},"subject":"Add autodocumentation for roles and library","message":"Add autodocumentation for roles and library\n\nThis patch adds automation to add documentation for each Ansible roles\nand custom modules for TripleO-Validations.\n\nPatch-Inspired-By: tripleo-ansible-Team++\nChange-Id: I2c8ec698cdcbc304b5d9f3bd72bcede5bf2aafd3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
