)]}'
{"id":"openstack%2Ftripleo-validations~672745","triplet_id":"openstack%2Ftripleo-validations~master~I3923fb89ac6e43de3e24b8c649dcfd30a72682b6","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework-infra","hashtags":[],"change_id":"I3923fb89ac6e43de3e24b8c649dcfd30a72682b6","subject":"Add Molecule tests for check-network-gateway","status":"MERGED","created":"2019-07-25 14:33:28.000000000","updated":"2019-08-28 10:27:23.000000000","submitted":"2019-08-28 10:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672745-1566988043406-1d226e13","meta_rev_id":"9e70e27ef9ad9b9c2a5793085f94687e5181f588","_number":672745,"virtual_id_number":672745,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-28 10:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"date":"2019-08-21 08:57:12.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-08-21 06:07:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-23 15:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-08-21 14:10:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-28 06:45:36.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"date":"2019-08-26 11:16:22.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 06:45:36.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 14:35:35.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-20 10:10:08.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-21 14:10:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-23 15:00:07.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-26 11:16:22.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-08-28 06:45:36.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-28 10:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bb6e5a0ba66991053aa073bb72ae71c487f0621","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-25 14:33:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d60e5e4e0bfde3caf3e0e407100ce8122659f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:19:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-containers-multinode/9712f10/ : SUCCESS in 2h 31m 50s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-undercloud-containers/a5821e1/ : SUCCESS in 1h 33m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/aa1aef0/ : SUCCESS in 2h 24m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/18a36d4/ : SUCCESS in 2h 22m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-standalone/e569a6b/ : SUCCESS in 1h 46m 55s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/45/672745/1/check/tripleo-ci-fedora-28-standalone/bd2aa91/ : SUCCESS in 1h 56m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/45/672745/1/check/tripleo-ci-centos-7-standalone-upgrade/4a726e8/ : FAILURE in 1h 44m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/45/672745/1/check/openstack-tox-pep8/0f5117e/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/45/672745/1/check/openstack-tox-py27/9456362/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/45/672745/1/check/openstack-tox-py36/6dfb642/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/45/672745/1/check/openstack-tox-py37/c3adf67/ : SUCCESS in 4m 53s\n- openstack-tox-molecule http://logs.openstack.org/45/672745/1/check/openstack-tox-molecule/d6e2159/tox/reports.html : TIMED_OUT in 31m 09s\n- openstack-tox-docs http://logs.openstack.org/45/672745/1/check/openstack-tox-docs/96ce0ed/html/ : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672745/1/check/openstack-tox-lower-constraints/56e5947/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5f0b5a558b9d0c4f89ccf8220e7b72bfcb66a16a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-26 05:21:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"872b8cfe88d36061d001bed66713df2943c65fe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 07:53:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-containers-multinode/9288840/ : SUCCESS in 2h 08m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-undercloud-containers/f495f25/ : SUCCESS in 1h 33m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0ebc7e9/ : SUCCESS in 2h 04m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cc63c66/ : SUCCESS in 2h 23m 56s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-standalone/fdd70a8/ : SUCCESS in 1h 44m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/45/672745/2/check/tripleo-ci-fedora-28-standalone/2e1de52/ : SUCCESS in 1h 45m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/45/672745/2/check/tripleo-ci-centos-7-standalone-upgrade/2328fc1/ : FAILURE in 2h 13m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/45/672745/2/check/openstack-tox-pep8/2cb087b/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/45/672745/2/check/openstack-tox-py27/9a0cb8b/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/45/672745/2/check/openstack-tox-py36/d10be04/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/45/672745/2/check/openstack-tox-py37/3888888/ : SUCCESS in 4m 13s\n- openstack-tox-molecule http://logs.openstack.org/45/672745/2/check/openstack-tox-molecule/01153aa/tox/reports.html : TIMED_OUT in 31m 10s\n- openstack-tox-docs http://logs.openstack.org/45/672745/2/check/openstack-tox-docs/8a48688/html/ : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672745/2/check/openstack-tox-lower-constraints/3c6a2b1/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c64ff72e142eea594590746e90473bebf1860ddc","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-30 06:46:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1ef5186f0a894ad3af0e15209844823053149a40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:09:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-containers-multinode/99d1a24/ : SUCCESS in 2h 18m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-undercloud-containers/453d70f/ : SUCCESS in 1h 29m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/014947b/ : SUCCESS in 2h 19m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f2d968b/ : FAILURE in 2h 32m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-standalone/905fd37/ : SUCCESS in 1h 48m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/45/672745/3/check/tripleo-ci-fedora-28-standalone/cd9314f/ : SUCCESS in 1h 50m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/45/672745/3/check/tripleo-ci-centos-7-standalone-upgrade/c787a31/ : SUCCESS in 2h 15m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/45/672745/3/check/openstack-tox-pep8/eec81ee/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/45/672745/3/check/openstack-tox-py27/97349e1/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/45/672745/3/check/openstack-tox-py36/ccc8945/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/45/672745/3/check/openstack-tox-py37/211fe5d/ : SUCCESS in 4m 29s\n- openstack-tox-molecule http://logs.openstack.org/45/672745/3/check/openstack-tox-molecule/daf9ec3/tox/reports.html : TIMED_OUT in 30m 59s\n- openstack-tox-docs http://logs.openstack.org/45/672745/3/check/openstack-tox-docs/b57b354/html/ : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672745/3/check/openstack-tox-lower-constraints/9270ccd/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"488e006c828489a5df55199ad2edff3204a52ae4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-04 12:19:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"80192620f957f691ff649b21e3aea4657f44d3c5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-04 12:58:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3373c47397eeb045d84d0cf6cad50319218af8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 15:33:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-containers-multinode/e0a354e/ : SUCCESS in 2h 28m 28s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-undercloud-containers/5278017/ : SUCCESS in 1h 39m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/963b687/ : SUCCESS in 2h 15m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5ec3db9/ : SUCCESS in 2h 27m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-standalone/5136bef/ : FAILURE in 50m 21s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/45/672745/5/check/tripleo-ci-fedora-28-standalone/42b24eb/ : FAILURE in 1h 04m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/45/672745/5/check/tripleo-ci-centos-7-standalone-upgrade/aa4b588/ : FAILURE in 39m 30s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/45/672745/5/check/openstack-tox-pep8/3621a22/ : SUCCESS in 4m 43s\n- openstack-tox-py27 https://logs.opendev.org/45/672745/5/check/openstack-tox-py27/b69b391/ : SUCCESS in 5m 06s\n- openstack-tox-py36 https://logs.opendev.org/45/672745/5/check/openstack-tox-py36/8495541/ : SUCCESS in 4m 12s\n- openstack-tox-py37 https://logs.opendev.org/45/672745/5/check/openstack-tox-py37/1937178/ : SUCCESS in 5m 30s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://logs.opendev.org/45/672745/5/check/tripleo-validations-centos-7-molecule-check-network-gateway/07765e9/reports.html : SUCCESS in 9m 16s\n- openstack-tox-linters https://logs.opendev.org/45/672745/5/check/openstack-tox-linters/8bf099c/ : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/672745/5/check/openstack-tox-lower-constraints/86fd8a3/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"73d0823a7aa634e033d60fdf8ee9cc0b78636c13","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 07:35:04.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"48dc3832320f799cc163e9170c0989f3524d7609","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 10:52:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2f33412fcc8b3f6e6e14f4eb3d138e99d9bc8d26","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 12:18:56.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"071b08c2a5f31410c15307c598ec599fdac1e6eb","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 13:21:43.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"25924e703d8ed0d3510a57a54564f9733cff7fe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 19:48:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-containers-multinode/1815d69/ : SUCCESS in 2h 32m 33s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-undercloud-containers/62aab71/ : SUCCESS in 1h 43m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/83ee8b8/ : SUCCESS in 2h 20m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3b33052/ : SUCCESS in 2h 28m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-standalone/87fe121/ : SUCCESS in 1h 58m 32s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/45/672745/9/check/tripleo-ci-fedora-28-standalone/7770bdc/ : FAILURE in 1h 31m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/45/672745/9/check/tripleo-ci-centos-7-standalone-upgrade/ea25bdf/ : TIMED_OUT in 3h 21m 18s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/45/672745/9/check/openstack-tox-pep8/78a3446/ : SUCCESS in 5m 03s\n- openstack-tox-py27 https://logs.opendev.org/45/672745/9/check/openstack-tox-py27/ec1eefc/ : SUCCESS in 4m 36s\n- openstack-tox-py36 https://logs.opendev.org/45/672745/9/check/openstack-tox-py36/6f1bcda/ : SUCCESS in 4m 26s\n- openstack-tox-py37 https://logs.opendev.org/45/672745/9/check/openstack-tox-py37/b5e0db0/ : SUCCESS in 5m 27s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://logs.opendev.org/45/672745/9/check/tripleo-validations-centos-7-molecule-check-network-gateway/3cae26b/reports.html : SUCCESS in 10m 54s\n- openstack-tox-linters https://logs.opendev.org/45/672745/9/check/openstack-tox-linters/f91b3ad/ : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/672745/9/check/openstack-tox-lower-constraints/b420712/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":9},{"id":"2cf2ec6e01c8c518b94acc58950e76ed37f2dfe4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-07 08:19:03.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"872a09797e261b57f327c1d72905d13be19c1185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:08:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-containers-multinode/48e3efb/ : SUCCESS in 2h 34m 01s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-undercloud-containers/b3d886a/ : SUCCESS in 1h 46m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2f46192/ : SUCCESS in 2h 26m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f93d2ac/ : SUCCESS in 2h 52m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-standalone/585621b/ : SUCCESS in 2h 07m 14s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/45/672745/10/check/tripleo-ci-fedora-28-standalone/f8d43f5/ : SUCCESS in 2h 00m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/45/672745/10/check/tripleo-ci-centos-7-standalone-upgrade/5e01c88/ : SUCCESS in 2h 12m 40s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/45/672745/10/check/openstack-tox-pep8/7dcaeb3/ : SUCCESS in 5m 36s\n- openstack-tox-py27 https://logs.opendev.org/45/672745/10/check/openstack-tox-py27/cf947e2/ : SUCCESS in 7m 50s\n- openstack-tox-py36 https://logs.opendev.org/45/672745/10/check/openstack-tox-py36/1d34f3e/ : SUCCESS in 4m 29s\n- openstack-tox-py37 https://logs.opendev.org/45/672745/10/check/openstack-tox-py37/89d3782/ : SUCCESS in 5m 22s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://logs.opendev.org/45/672745/10/check/tripleo-validations-centos-7-molecule-check-network-gateway/0cb4685/reports.html : SUCCESS in 11m 42s\n- openstack-tox-linters https://logs.opendev.org/45/672745/10/check/openstack-tox-linters/7005e22/ : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/672745/10/check/openstack-tox-lower-constraints/4103743/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":10},{"id":"1d2524b9748ddbb0dee93130caab9c164bc055f8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-09 07:25:35.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"d8c15737dacd5f4865479f7b5bb50ff6075ecc09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 10:43:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-containers-multinode/17d2d61/ : SUCCESS in 2h 04m 35s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-undercloud-containers/c052da4/ : SUCCESS in 1h 33m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5443e42/ : SUCCESS in 2h 13m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/02550d8/ : SUCCESS in 2h 20m 09s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-standalone/fd25da9/ : SUCCESS in 1h 42m 17s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/45/672745/11/check/tripleo-ci-fedora-28-standalone/597aa3f/ : SUCCESS in 1h 59m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/45/672745/11/check/tripleo-ci-centos-7-standalone-upgrade/f248c4f/ : SUCCESS in 2h 17m 41s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/45/672745/11/check/openstack-tox-pep8/5a9cfee/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/45/672745/11/check/openstack-tox-py27/c8c0e78/ : SUCCESS in 4m 28s\n- openstack-tox-py36 https://logs.opendev.org/45/672745/11/check/openstack-tox-py36/74b3bdb/ : SUCCESS in 5m 13s\n- openstack-tox-py37 https://logs.opendev.org/45/672745/11/check/openstack-tox-py37/f22a05f/ : SUCCESS in 4m 59s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://logs.opendev.org/45/672745/11/check/tripleo-validations-centos-7-molecule-check-network-gateway/73963ef/reports.html : SUCCESS in 10m 57s\n- openstack-tox-linters https://logs.opendev.org/45/672745/11/check/openstack-tox-linters/c4e7581/ : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/45/672745/11/check/openstack-tox-lower-constraints/9fc4d5b/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":11},{"id":"7d0bc15b8d05ce09e29de4cef22a263e9e9447fe","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:46:06.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"aa8c3ea3d1607008398ff5f2cabb790ead533968","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 10:56:54.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"8378216a1075149df287c4eb9d8003b2a05ba9e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 10:57:08.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"7b9bc32eafa09ffdb9da07f169d8854cb1edb6dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:53:40.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ef323d732c34442db8e54420f48d3c90 : SUCCESS in 2h 54m 39s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4bd586cdeb724f17a347bb0cefba8311 : FAILURE in 1h 00m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4f16016863964e45bf2968730a2556eb : SUCCESS in 2h 35m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843461f3e7bb472a97c6823f8345d30c : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/845bb180484046c5807e909b17923717 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e9dfe3c472f43018eab0b1cc9870684 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0be55921ef514f72a3112e8b7c3514e5 : SUCCESS in 9m 13s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/eb2b2103f279456b8626be118791c01f : SUCCESS in 13m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f1bff63c92f0435089e2d2800b20c0a3 : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce05856119cc4cf394bfd330b6a9e4e2 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":11},{"id":"ee0b60d62f4ecf6af2e47df97f63d40c4f9b40cd","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 14:02:49.000000000","message":"Patch Set 11: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"dd641dedc8176074ef37b93dc689c1184551bec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:50:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/378eb8e4fed9442990438d231e2daa4c : SUCCESS in 2h 43m 10s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/de1684e4831b4ccda3696c78314fa2cf : SUCCESS in 1h 50m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/946eb10cc729475bafb21ce530170848 : FAILURE in 2h 28m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/054577f5b143442691063cef53edb85e : SUCCESS in 2h 42m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6a70eb20fd85440d819ff2b6b197e737 : SUCCESS in 1h 58m 54s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6574025a77b24360828a07e9c1923375 : SUCCESS in 2h 11m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b125ce5604483da94e59c4895fbd59 : SUCCESS in 7m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfe94f3c37a74ec3b1e3b4c6834e1067 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369e9cbce7894a7caa85058c8d8309cb : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1367b57d86c24b55ad189a99a02af481 : SUCCESS in 5m 13s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/8aa12698557b4652a3112e7edb51d97c : SUCCESS in 10m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/afedde77fbf543c5b9f9cd679808aebe : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deb0938745f0414385108078e2f63666 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":11},{"id":"e8a7c7c4924c07d4a24bc08fdaa4556f55a474af","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-20 10:10:08.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"dd2e81273ced20f9c03eb285ed6c1521c204d821","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-20 10:10:13.000000000","message":"Removed Code-Review+2 by Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"01f8984f15e3c343fd8be187152bc37e6575bf49","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-20 10:10:15.000000000","message":"Removed Workflow+1 by Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"b6d6eaf248db31125e928a11e1681fef29cabcb5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-20 12:10:26.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"5bcb1e210ca1cd1232ed99cf521aeadde82ea909","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 12:56:58.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/67ef73a4bdf849b8a67ebc41219f3e2c : FAILURE in 2h 44m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c1ebab691bc04eb392d17d4b67c02354 : SUCCESS in 1h 56m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/113bdfed7de74ab6895574764eb3f163 : SUCCESS in 2h 24m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d2c8e183a42b4557adc4b453265621ee : SUCCESS in 2h 30m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8acc54acaf8f40a491fa290b70c336c6 : SUCCESS in 2h 02m 21s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/af189afb51ba4fedb07b0bd9716f1a0a : SUCCESS in 2h 15m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab5c33b080a4bb089885b0f32357b6c : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acbe954b74b44751b4530d73a99c3f6b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c15b05fefde40dbb757b63ff6222142 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/434e7527c7024b70aa1c285f7669f543 : SUCCESS in 4m 42s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/910db698118a4b898cfca8a709173aff : SUCCESS in 12m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ffc896143384cdb978452e27fd0d9db : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf9a5b700f934c0198557f9502db35df : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":11},{"id":"0d0cf41738c9483710348c35c9492e5baa7d23b2","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-20 14:31:03.000000000","message":"Patch Set 11:\n\nrecheck\n....","accounts_in_message":[],"_revision_number":11},{"id":"5c928a9b8063a361543680ae77b6c1ed4f644fcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:38:46.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c916289019f24a36b36596ec72811e1a : FAILURE in 15m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/466b1274bab64c4b92dba8f3c1a72e40 : SUCCESS in 1h 52m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/82b0ab2d6ce14fd6bf4845f40d216fd9 : SUCCESS in 2h 44m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1b3c3051525847d88816878e389964d4 : SUCCESS in 2h 54m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1f84db6602f242809cf4da04d5a0feb0 : SUCCESS in 1h 55m 17s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/74b00183f8324b578ba710a364f65ec6 : SUCCESS in 2h 17m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b190b16de74431a84af9632182da7b9 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f723dadf684138b4e5512f1b7a8d9a : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21685680454441eaa5170a33e62e3401 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c63342be13e47fab6cb5e786fab23ba : SUCCESS in 5m 46s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/3334e93dba184c2fb017db2f49f3d604 : SUCCESS in 11m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bf5ccd56a67948649296b68ea905cbda : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1730a734c2fa4236ac8384ebeec2938c : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":11},{"id":"d4e5e2d53e31c62a55bc6c2133c38b777ded7275","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-21 05:29:15.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"1319be0df968fe455b5d2a6d156b9224a553a9e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 05:56:29.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2a0213e739b746708c49d8adf8bf80cb : FAILURE in 14m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e27046a1e14c4b5699b2d70a4dadb0fd : FAILURE in 11m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f00bb7edf42d4c628f8527a280e4ab0c : FAILURE in 11m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/89a8cb68ce9a4305bf145979d4be8a89 : FAILURE in 12m 37s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/103c9c02a742488bb3f4b3ea7c5332ab : FAILURE in 6m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/24e296fd92154f86a2000ff1f3a8d838 : FAILURE in 9m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1777e803c294e9b860a8affa68af96e : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd2a8eaadbe64009b58c787891998cbb : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a396362b47c341c7a5658f047bc4ee34 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3397c1e76e7849a38d764f0a64f57d79 : SUCCESS in 5m 08s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/127e16e981eb43219fbc3e56e7f6bee8 : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/404d33943267443fa141d4e455e4d611 : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29eba92d13414095a47a8dad7935a0fa : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":11},{"id":"c88152dfe7b042c71fdbeabf79af82c287679fdd","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-21 06:07:09.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"97586ca3e4a2fa1acb59d57111d86a52e4856387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:17:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b195cd993a6041a9abc42f749faa0d6e : FAILURE in 12m 02s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eae9aa1e2c5c4884ba6a658aad9c5429 : FAILURE in 8m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8d223e260ec4403aadfd535f934b0126 : FAILURE in 12m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/88d3ffa47bfb4398a189064b1bf90457 : FAILURE in 18m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d1fbab26c8024796b6cac224e6bcd9de : FAILURE in 15m 39s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd7e9d2e275c41f5a8b2dea605a672e8 : FAILURE in 13m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e032770d384f4123a13124f6cfa24b2c : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96832c4e275a444d866ad49ca2e8064d : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b5f94653a4d4968a8f486afaf76ef20 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1d83f1d656942949769f204ee2dc23b : SUCCESS in 7m 32s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/7e05fcbb1f094316ac5ae363b496a478 : SUCCESS in 12m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4a832b754be4c2ca66e69aa4937180f : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de4bd1c2da294c9db55e32d3f08c7231 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":12},{"id":"d9315cf0063355413679ea7ca53a7278adb9ded5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-21 08:57:12.000000000","message":"Patch Set 12:\n\nrecheck\nhttps://review.opendev.org/#/c/677669/ fixed the issue.","accounts_in_message":[],"_revision_number":12},{"id":"7f41f74754570b08c2f68ae3909dfb3513491de9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 11:50:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/75226cbd2303492881c1c9f306302c51 : SUCCESS in 2h 52m 05s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cef28a5f545c486e8ad6c4f726ce467b : SUCCESS in 1h 59m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cf003e0172484977b69697c9c8fef8ed : SUCCESS in 2h 46m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/aa3fd467292d4614815d5775c0e88be7 : FAILURE in 17m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fb779e8e6ea0480f99fa3c8c7161bba6 : SUCCESS in 2h 01m 28s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/32ddb97533e94049a9021e49c5b13b39 : SUCCESS in 2h 07m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39afff0726ae4b0fb6d674930a50b73a : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c369bd9c32a4ea78530958980a87ee4 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e65c3826d01d4d6fb30b7e060c223177 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6621432fc16540ee8a54b64e61ac0d62 : SUCCESS in 5m 24s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/4361fd9dfa0a46a393101f67c9215b05 : SUCCESS in 10m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59a1fd6cd727433293e9e269f42e24ce : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03e8c69ee55b45d99bf15e7461d7a16d : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":12},{"id":"3b2018ae90b7313ffb1028e2b43d8bdd1e497136","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-21 14:10:56.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"a9917affb56b6adc9d54bf7ce687bf4675c290d3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-23 15:00:07.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"21161072fad5c9996cf3b3946e685cd9112d2c40","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-28 06:45:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"bc6cb8c5f807bed33d88fe612f5ef17e3464e579","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 06:45:51.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e583b688bb3e1987cf15f551950503e90009573e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:27:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"2bd71f327303bf93f20925ea3c2acc47b741d10d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:27:23.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bc4d54f667a94a9bb96ed3d1925ae4ad : SUCCESS in 3h 00m 22s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/28970dc04d454861a2140d004833e778 : SUCCESS in 2h 04m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1ce47dbdc3804e3f81616e22ea0c4eb5 : SUCCESS in 3h 03m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92605cb08244921a76bea9e3d178e52 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84f0598404784700a51eca41134a02de : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d0dcad49e14221b4b61bb98f5da83b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d698e507af7c48f08e135fe07287f177 : SUCCESS in 5m 21s\n- tripleo-validations-centos-7-molecule-check-network-gateway https://zuul.opendev.org/t/openstack/build/5a956f3cd47e42bea5a539b236dec39f : SUCCESS in 10m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36112c74abe140e781a09f2ac9334de2 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb3b9d22c30642b7bdf90954ebbf566f : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6dffc61d946ae220855a950dde4c9a3aee071933","revisions":{"8b6f0a37d84b75d423bffd4097009f4b9da8d09f":{"kind":"REWORK","_number":1,"created":"2019-07-25 14:33:28.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/45/672745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/1"}}},"commit":{"parents":[{"commit":"680512410fdd93d2a3f2b23e570c90fcad7ff8ea","subject":"Merge \"Adds molecule tests for ctlplane-ip-range validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/680512410fdd93d2a3f2b23e570c90fcad7ff8ea"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b6f0a37d84b75d423bffd4097009f4b9da8d09f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b6f0a37d84b75d423bffd4097009f4b9da8d09f"}]},"branch":"refs/heads/master"},"c8a4726167e3a22ea1a7e2c8d9fed6a935297080":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-26 05:21:52.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/45/672745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/2"}}},"commit":{"parents":[{"commit":"13e62cd2be4012e9d5fa33c8302e0ac0b44948e7","subject":"Adds molecule tests to image-server and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/13e62cd2be4012e9d5fa33c8302e0ac0b44948e7"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 05:21:45.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c8a4726167e3a22ea1a7e2c8d9fed6a935297080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c8a4726167e3a22ea1a7e2c8d9fed6a935297080"}]},"branch":"refs/heads/master"},"88df5805214035a7acae46b3a89928f55ff8ec33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-30 06:46:35.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/45/672745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/3"}}},"commit":{"parents":[{"commit":"d9ce4d92624e72519fd5695f7c4be55ab2e4cbb1","subject":"Adds molecule tests to image-server and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d9ce4d92624e72519fd5695f7c4be55ab2e4cbb1"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-30 06:45:13.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88df5805214035a7acae46b3a89928f55ff8ec33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88df5805214035a7acae46b3a89928f55ff8ec33"}]},"branch":"refs/heads/master"},"7780b7d0c78eee813b820c262f74a85b3d0155c7":{"kind":"REWORK","_number":4,"created":"2019-08-04 12:19:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/4"}}},"commit":{"parents":[{"commit":"d0071c2505938d5bf28efe0f85f7f97d344e4331","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d0071c2505938d5bf28efe0f85f7f97d344e4331"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-04 12:08:25.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7780b7d0c78eee813b820c262f74a85b3d0155c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7780b7d0c78eee813b820c262f74a85b3d0155c7"}]},"branch":"refs/heads/master"},"b2e6c121650835dd4736072ade341006e3325b45":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-04 12:58:13.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/5"}}},"commit":{"parents":[{"commit":"5ea693843c47b0e86423279d888e00723293fc18","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5ea693843c47b0e86423279d888e00723293fc18"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-04 12:57:50.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b2e6c121650835dd4736072ade341006e3325b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b2e6c121650835dd4736072ade341006e3325b45"}]},"branch":"refs/heads/master"},"6eb1d1f878d4f179f45b972020756afb0045cdb3":{"kind":"NO_CHANGE","_number":6,"created":"2019-08-06 07:35:04.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/6"}}},"commit":{"parents":[{"commit":"541735d0a90857daa249dac38f4699c6aaaa8264","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/541735d0a90857daa249dac38f4699c6aaaa8264"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 07:34:28.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6eb1d1f878d4f179f45b972020756afb0045cdb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6eb1d1f878d4f179f45b972020756afb0045cdb3"}]},"branch":"refs/heads/master"},"9beb438b6f690d62e6ba1757a63e323b7c2b1d3f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-06 10:52:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/7"}}},"commit":{"parents":[{"commit":"9335d808da689e3813dcfb5214991d4fc9708e8c","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9335d808da689e3813dcfb5214991d4fc9708e8c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 10:51:43.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9beb438b6f690d62e6ba1757a63e323b7c2b1d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9beb438b6f690d62e6ba1757a63e323b7c2b1d3f"}]},"branch":"refs/heads/master"},"e355b06183c82f7a0aa57b3369dfa49195ad3eff":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-06 12:18:56.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/8"}}},"commit":{"parents":[{"commit":"779735561037a86a7eeddb6d190d505fa6467b02","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/779735561037a86a7eeddb6d190d505fa6467b02"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 12:14:57.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e355b06183c82f7a0aa57b3369dfa49195ad3eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e355b06183c82f7a0aa57b3369dfa49195ad3eff"}]},"branch":"refs/heads/master"},"acdeb520bcd43798dc8f3cf23f9396c3b85b93a6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-06 13:21:43.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/9"}}},"commit":{"parents":[{"commit":"f58f3ee3e2254a4221bc2b6df7fadad4da7f36ec","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f58f3ee3e2254a4221bc2b6df7fadad4da7f36ec"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 13:21:32.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/acdeb520bcd43798dc8f3cf23f9396c3b85b93a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/acdeb520bcd43798dc8f3cf23f9396c3b85b93a6"}]},"branch":"refs/heads/master"},"d194621531449590a46b4665ea53293b006d168e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-07 08:19:03.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/10"}}},"commit":{"parents":[{"commit":"86880d640ab45994916a46c7c8205d62dbdd36f5","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/86880d640ab45994916a46c7c8205d62dbdd36f5"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-07 08:17:08.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d194621531449590a46b4665ea53293b006d168e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d194621531449590a46b4665ea53293b006d168e"}]},"branch":"refs/heads/master"},"85ca822239be4930998e4f3c7c8154016be4f929":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-08-09 07:25:35.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/45/672745/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/11"}}},"commit":{"parents":[{"commit":"6ae16f588e60e7e043380641eb15a18f5dd1b854","subject":"Adds molecule tests to image-serve and correct validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6ae16f588e60e7e043380641eb15a18f5dd1b854"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-09 07:24:17.000000000","tz":120},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/85ca822239be4930998e4f3c7c8154016be4f929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/85ca822239be4930998e4f3c7c8154016be4f929"}]},"branch":"refs/heads/master"},"6dffc61d946ae220855a950dde4c9a3aee071933":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-08-21 06:07:09.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/45/672745/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/45/672745/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/45/672745/12"}}},"commit":{"parents":[{"commit":"6a8f3141b11165f3d4360e838441ea88ce6bcc97","subject":"Merge \"Removed older version of python\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6a8f3141b11165f3d4360e838441ea88ce6bcc97"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-25 14:32:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret (Tengu)","email":"cjeanner@redhat.com","date":"2019-08-21 06:07:09.000000000","tz":0},"subject":"Add Molecule tests for check-network-gateway","message":"Add Molecule tests for check-network-gateway\n\nChange-Id: I3923fb89ac6e43de3e24b8c649dcfd30a72682b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6dffc61d946ae220855a950dde4c9a3aee071933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6dffc61d946ae220855a950dde4c9a3aee071933"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
