)]}'
{"id":"openstack%2Ftripleo-validations~672889","triplet_id":"openstack%2Ftripleo-validations~master~I26275ce12bbb824a24a7d2571d1920208cd34aca","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework-infra","hashtags":[],"change_id":"I26275ce12bbb824a24a7d2571d1920208cd34aca","subject":"Adds molecule testing for undercloud-tokenflush","status":"MERGED","created":"2019-07-26 07:04:01.000000000","updated":"2019-08-28 21:18:49.000000000","submitted":"2019-08-28 21:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672889-1567027129712-5a467fae","meta_rev_id":"43858af378575d001b4208ced426462857248826","_number":672889,"virtual_id_number":672889,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-28 21:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-28 12:01:52.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-22 17:57:01.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-23 15:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-08-27 21:40:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-26 11:16:36.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-27 21:40:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 07:04:35.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-26 07:04:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-19 08:46:24.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-23 15:04:06.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-26 11:16:36.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-08-27 21:40:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-28 21:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ba414608465e124e2652335a71f41e5f0cbd331","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-26 07:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdfdefc025637aae58299f279d0c8dbb17847d09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:53:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-containers-multinode/39cc9ae/ : SUCCESS in 2h 15m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-undercloud-containers/e6f6b2f/ : SUCCESS in 1h 29m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3bb7489/ : SUCCESS in 2h 16m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f8b04f0/ : SUCCESS in 2h 32m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-standalone/e591e71/ : SUCCESS in 1h 54m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/89/672889/1/check/tripleo-ci-fedora-28-standalone/3a36ea4/ : SUCCESS in 1h 53m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/89/672889/1/check/tripleo-ci-centos-7-standalone-upgrade/8096ca3/ : FAILURE in 2h 11m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/672889/1/check/openstack-tox-pep8/4120f6e/ : SUCCESS in 7m 17s\n- openstack-tox-py27 http://logs.openstack.org/89/672889/1/check/openstack-tox-py27/4107cd6/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/89/672889/1/check/openstack-tox-py36/4f5b405/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/89/672889/1/check/openstack-tox-py37/f57d896/ : SUCCESS in 4m 23s\n- openstack-tox-molecule http://logs.openstack.org/89/672889/1/check/openstack-tox-molecule/624aeed/tox/reports.html : TIMED_OUT in 31m 01s\n- openstack-tox-docs http://logs.openstack.org/89/672889/1/check/openstack-tox-docs/5ed3533/html/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/672889/1/check/openstack-tox-lower-constraints/2b26128/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"078ee46cb07ca4a12f5e6caeba0d33cee1c739b5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-30 06:46:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9ca803bedac65902be97dc9f321796822c57cca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:04:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-containers-multinode/25d00db/ : SUCCESS in 2h 22m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-undercloud-containers/03c2254/ : SUCCESS in 1h 41m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1757c45/ : SUCCESS in 2h 21m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/20b4b7d/ : SUCCESS in 2h 29m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-standalone/3860f02/ : SUCCESS in 2h 05m 35s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/89/672889/2/check/tripleo-ci-fedora-28-standalone/efcbe9b/ : SUCCESS in 1h 55m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/89/672889/2/check/tripleo-ci-centos-7-standalone-upgrade/b361a17/ : SUCCESS in 2h 00m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/89/672889/2/check/openstack-tox-pep8/14b3aa6/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/89/672889/2/check/openstack-tox-py27/7a62792/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/89/672889/2/check/openstack-tox-py36/f0ec808/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/89/672889/2/check/openstack-tox-py37/598670e/ : SUCCESS in 6m 04s\n- openstack-tox-molecule http://logs.openstack.org/89/672889/2/check/openstack-tox-molecule/07f39f2/tox/reports.html : TIMED_OUT in 31m 37s\n- openstack-tox-docs http://logs.openstack.org/89/672889/2/check/openstack-tox-docs/b387ca8/html/ : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/672889/2/check/openstack-tox-lower-constraints/4d2c30b/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b72cbda324ce15eb41419520210d57f77dae7f61","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-04 12:19:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d19d44365c2911493e09e1a7b81240185da555a3","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-04 12:58:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a4eb22a1e9350fd3ca36ec3b10e787c0c8a12a1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 15:28:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-containers-multinode/620659a/ : SUCCESS in 2h 16m 38s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-undercloud-containers/d1304bb/ : SUCCESS in 1h 39m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ded3160/ : SUCCESS in 2h 14m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d36ba81/ : SUCCESS in 2h 26m 28s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-standalone/52bf4b0/ : FAILURE in 48m 43s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/89/672889/4/check/tripleo-ci-fedora-28-standalone/da6c98c/ : FAILURE in 54m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/89/672889/4/check/tripleo-ci-centos-7-standalone-upgrade/4c8e4cc/ : FAILURE in 36m 52s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/89/672889/4/check/openstack-tox-pep8/9d4966f/ : SUCCESS in 4m 39s\n- openstack-tox-py27 https://logs.opendev.org/89/672889/4/check/openstack-tox-py27/b4ff0c4/ : SUCCESS in 4m 32s\n- openstack-tox-py36 https://logs.opendev.org/89/672889/4/check/openstack-tox-py36/b8a2406/ : SUCCESS in 4m 48s\n- openstack-tox-py37 https://logs.opendev.org/89/672889/4/check/openstack-tox-py37/223425e/ : SUCCESS in 5m 01s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://logs.opendev.org/89/672889/4/check/tripleo-validations-centos-7-molecule-undercloud-tokenflush/b66557c/reports.html : SUCCESS in 10m 11s\n- openstack-tox-linters https://logs.opendev.org/89/672889/4/check/openstack-tox-linters/a3fe22d/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://logs.opendev.org/89/672889/4/check/openstack-tox-lower-constraints/51061bb/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"527529c6270e4d82e7489ab1e3c26e2e44024777","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 07:35:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2e73ff209a88bbe69ece1f60e90453052026153a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 10:52:34.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"45a4f155183f26ebabdd4998cb1495a1749b860b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 12:18:56.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"f3bcee03463bf25069eb0f4fb1811fb9bc3c75ad","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 13:21:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"52e5913d9d3872f242424a773ec6a0f57df389cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 19:16:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-containers-multinode/e7ef3f3/ : SUCCESS in 2h 30m 26s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-undercloud-containers/926eff6/ : SUCCESS in 1h 46m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/56fe0d8/ : SUCCESS in 2h 24m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b636582/ : SUCCESS in 2h 45m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-standalone/a032052/ : SUCCESS in 1h 47m 08s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/89/672889/8/check/tripleo-ci-fedora-28-standalone/938dc52/ : SUCCESS in 1h 46m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/89/672889/8/check/tripleo-ci-centos-7-standalone-upgrade/1d702d4/ : SUCCESS in 2h 13m 16s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/89/672889/8/check/openstack-tox-pep8/70a7951/ : SUCCESS in 5m 46s\n- openstack-tox-py27 https://logs.opendev.org/89/672889/8/check/openstack-tox-py27/68033ba/ : SUCCESS in 5m 27s\n- openstack-tox-py36 https://logs.opendev.org/89/672889/8/check/openstack-tox-py36/77d597d/ : SUCCESS in 4m 13s\n- openstack-tox-py37 https://logs.opendev.org/89/672889/8/check/openstack-tox-py37/245a7c0/ : SUCCESS in 5m 56s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://logs.opendev.org/89/672889/8/check/tripleo-validations-centos-7-molecule-undercloud-tokenflush/3337f7e/reports.html : SUCCESS in 10m 12s\n- openstack-tox-linters https://logs.opendev.org/89/672889/8/check/openstack-tox-linters/4acfc0e/ : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/89/672889/8/check/openstack-tox-lower-constraints/1276ad1/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":8},{"id":"1663b6ee640673690041bafbababd72a2ca57be9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-07 08:19:03.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b1f1a9f5a12bb7e52d6f717be1ad24e8c8e70ed5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:20:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-containers-multinode/659f8e0/ : SUCCESS in 2h 12m 07s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-undercloud-containers/758f58f/ : SUCCESS in 2h 12m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f15fc58/ : SUCCESS in 2h 27m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d74d3cb/ : SUCCESS in 3h 03m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-standalone/2cdb6ba/ : SUCCESS in 2h 01m 10s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/89/672889/9/check/tripleo-ci-fedora-28-standalone/2226de6/ : SUCCESS in 1h 44m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/89/672889/9/check/tripleo-ci-centos-7-standalone-upgrade/70a9ced/ : SUCCESS in 2h 09m 11s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/89/672889/9/check/openstack-tox-pep8/0699063/ : SUCCESS in 5m 19s\n- openstack-tox-py27 https://logs.opendev.org/89/672889/9/check/openstack-tox-py27/4f0ef98/ : SUCCESS in 5m 47s\n- openstack-tox-py36 https://logs.opendev.org/89/672889/9/check/openstack-tox-py36/9a3dda9/ : SUCCESS in 4m 20s\n- openstack-tox-py37 https://logs.opendev.org/89/672889/9/check/openstack-tox-py37/90ee78f/ : SUCCESS in 4m 20s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://logs.opendev.org/89/672889/9/check/tripleo-validations-centos-7-molecule-undercloud-tokenflush/bfd1640/reports.html : SUCCESS in 10m 13s\n- openstack-tox-linters https://logs.opendev.org/89/672889/9/check/openstack-tox-linters/586c6f6/ : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://logs.opendev.org/89/672889/9/check/openstack-tox-lower-constraints/c3e6967/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":9},{"id":"21730243845e79ac3d9cbb95790d245b3c48876d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-09 07:25:35.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"f68d7d68f19dd63a3eef12448e0d734475cb992e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:35:25.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-containers-multinode/c2a65a3/ : SUCCESS in 3h 09m 03s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-undercloud-containers/1938880/ : SUCCESS in 1h 38m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c5f8b54/ : SUCCESS in 2h 19m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/33fd27a/ : SUCCESS in 2h 36m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-standalone/b7a213b/ : SUCCESS in 2h 04m 11s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/89/672889/10/check/tripleo-ci-fedora-28-standalone/a057453/ : SUCCESS in 1h 43m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/89/672889/10/check/tripleo-ci-centos-7-standalone-upgrade/fe9fd92/ : SUCCESS in 2h 13m 36s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/89/672889/10/check/openstack-tox-pep8/06c4e94/ : SUCCESS in 6m 32s\n- openstack-tox-py27 https://logs.opendev.org/89/672889/10/check/openstack-tox-py27/b9396c3/ : SUCCESS in 4m 15s\n- openstack-tox-py36 https://logs.opendev.org/89/672889/10/check/openstack-tox-py36/6d9f5dc/ : SUCCESS in 6m 45s\n- openstack-tox-py37 https://logs.opendev.org/89/672889/10/check/openstack-tox-py37/d8e8794/ : SUCCESS in 6m 11s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://logs.opendev.org/89/672889/10/check/tripleo-validations-centos-7-molecule-undercloud-tokenflush/89b004e/reports.html : SUCCESS in 11m 34s\n- openstack-tox-linters https://logs.opendev.org/89/672889/10/check/openstack-tox-linters/f7a87f3/ : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://logs.opendev.org/89/672889/10/check/openstack-tox-lower-constraints/c52c89c/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":10},{"id":"dc04f385f65f94614e14866cf0e0ba22150fcb37","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:46:24.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"184fcea72f843c0ef2b0db78851844d3168af9d4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 10:57:05.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"b0803ea98ca52368972deaf9c23b48879046991e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 10:57:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a8c00478c83771f5dd7abd2b1974c955af7bb7b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:53:41.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"1fd9512e2fd7904680fe402862bf0e1ca09e5533","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 17:57:01.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"5db858965888a0eba9fa3d613b1d2fb931ddb35d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:17:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cea78a5f96ef44bd8ef456101fb9cc8f : SUCCESS in 2h 37m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/707044df6a8e441fa0d1f1da91d24622 : SUCCESS in 1h 38m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5cec6121596a43e8ac54ae46eb52f5a5 : SUCCESS in 2h 38m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b77b848caac040ea90c8a26ac2e96245 : SUCCESS in 3h 15m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b0122bcbb54b42269b468830b828cb6a : SUCCESS in 1h 59m 39s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6a485507d576475c977c91ba094f11e5 : SUCCESS in 2h 27m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99b40f78c6442eb9674d5635527d24d : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ee5bb2a6a154caa9f859a42b4cf9c88 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1646ff5ed444b29cb2851c3405ce02 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1fe9e6f8ccc44a185a4d920ea854448 : SUCCESS in 5m 39s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://zuul.opendev.org/t/openstack/build/d21b69ff000d430f9eeab59d46de65ce : SUCCESS in 10m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/507b2c9e5e4b454a961940efa06e8673 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d57f99e8ebb401b900855eae06b6409 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":11},{"id":"41f49fffb1f36ca6eb5461724d9bec70d0e43034","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-23 15:04:06.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"a2dd3ad0d5bb01721097079d87678b058e35559a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-27 21:40:24.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"23391afa2da38aeff70e3d16ea1a9c4531c0287c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 06:45:58.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"fe28fce8e1bdad53f5a4fb13d51b8b3fc30acbd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:27:25.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4671c6e1a6724c00abd8c3c64054c23a : SUCCESS in 2h 47m 04s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ad6ee28b481346b7b480a260709279f0 : SUCCESS in 2h 08m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d3020a1500b643d988135422066f8c75 : SUCCESS in 2h 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ae1d209f4c4a8f878bb616ae176fac : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fb707b67c1545249a7f4104eccf0c5f : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7639797113b413397b55b7a8a3f8f02 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/365b654938a04d0ca28a0175bfa02ab6 : SUCCESS in 4m 56s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://zuul.opendev.org/t/openstack/build/d3b84c414fd24366ba86200362486042 : FAILURE in 8m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23284c3ddeb449b3a6f77f707332cadd : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9564312fafd34d95980211ae438eac05 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":11},{"id":"f26bd0cae9090fd8cce5cbdec1f2c92345993fcb","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-28 12:01:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"7ab449624b1f076ef6dc54f00027018b8a58063e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 15:01:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2e6f6ec0e9f6440a9231bd2a40f39af0 : SUCCESS in 2h 46m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4d8148c5d91c4d01b3aaf2c6e4e32210 : SUCCESS in 1h 48m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6715a4f406f14e10a6ea01ed418e5f4f : SUCCESS in 2h 52m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0dbd6c80d4a5481292ce2ba070c6337f : SUCCESS in 2h 54m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f86af4d4e87740ef932de6264f977c18 : SUCCESS in 2h 12m 52s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d88426aca383489cbf3ef1e24e7563a2 : SUCCESS in 2h 12m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c23080cd1054de6ba714b3f7ec8f7ff : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46eccfc8e5354772a9d0a8f40f133714 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8dccff0e5d446ab12ffc0c4d14db5a : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a286f417d2f84ca3afb1a484bb7b63d2 : SUCCESS in 8m 56s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://zuul.opendev.org/t/openstack/build/3c5520d3d50f4cd2a8e57bffc7f3dfe2 : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/451c9c8bc3fd44f18ff04bb3b97d2d53 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":11},{"id":"7d5aeeeafbeab08e37c62cc25c623309bf7c1070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 15:02:05.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"2d526b6cc29dd38fd5079b79e8e6d71716cab666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:18:49.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6df352304a3e48d9a75cf97017cc4a6d : SUCCESS in 3h 03m 34s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b58a754f44bd4bc1b5b0e86c345a5178 : SUCCESS in 2h 13m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/110539542a904f31a41672ca9a6a91af : SUCCESS in 2h 52m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af718c44460e49f5936e4b93ab90d55c : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f46abd1b4b54cb8b93922a88ddd40e7 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9b98933e62f4726a82c2ac6f57252d3 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b80a52c2be444c5d88a3039ca7302c1f : SUCCESS in 6m 56s\n- tripleo-validations-centos-7-molecule-undercloud-tokenflush https://zuul.opendev.org/t/openstack/build/05b57b216f4942e090bce57857461325 : SUCCESS in 10m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f2c8da518dc14d96ad581bcfac3df694 : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e9c5c6c60974830bd3c9aa7d1703072 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":11},{"id":"c1ca4c1ee25fcab4bea45db66c1d9f14f33e9af3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:18:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0cb6d6172ea078f4657594dc34c744a03c0420cd","revisions":{"ed4face14fc346edf368d5902db4c1a5c5681a3e":{"kind":"REWORK","_number":1,"created":"2019-07-26 07:04:01.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/89/672889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/1"}}},"commit":{"parents":[{"commit":"4ae720cd962c883666d748b3649cd84ed961ef36","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4ae720cd962c883666d748b3649cd84ed961ef36"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ed4face14fc346edf368d5902db4c1a5c5681a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ed4face14fc346edf368d5902db4c1a5c5681a3e"}]},"branch":"refs/heads/master"},"13a327702b251ad392a4aa019bd4d393fd3922d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-30 06:46:35.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/89/672889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/2"}}},"commit":{"parents":[{"commit":"53336740ed83a52a6b21179c80244ff6c89421b1","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/53336740ed83a52a6b21179c80244ff6c89421b1"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-30 06:45:52.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/13a327702b251ad392a4aa019bd4d393fd3922d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/13a327702b251ad392a4aa019bd4d393fd3922d3"}]},"branch":"refs/heads/master"},"95519e1758451bd48d236e15802e8c3fd3173a3c":{"kind":"REWORK","_number":3,"created":"2019-08-04 12:19:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/3"}}},"commit":{"parents":[{"commit":"f940c6f3ae737a1df49b74408d2b242d4bf23f14","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f940c6f3ae737a1df49b74408d2b242d4bf23f14"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-04 12:11:41.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/95519e1758451bd48d236e15802e8c3fd3173a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/95519e1758451bd48d236e15802e8c3fd3173a3c"}]},"branch":"refs/heads/master"},"316042704a4a2372cb27b87b40aad4b57bd97867":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-04 12:58:13.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/4"}}},"commit":{"parents":[{"commit":"36085caebff1739fc358e34ebdc4002b28a6e944","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/36085caebff1739fc358e34ebdc4002b28a6e944"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-04 12:57:50.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/316042704a4a2372cb27b87b40aad4b57bd97867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/316042704a4a2372cb27b87b40aad4b57bd97867"}]},"branch":"refs/heads/master"},"f93a6682dd2f1121608a8df2eba198da62160015":{"kind":"NO_CHANGE","_number":5,"created":"2019-08-06 07:35:04.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/5"}}},"commit":{"parents":[{"commit":"e8cbf791b16b5dcf165e25c81b911f5ae4c0369e","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e8cbf791b16b5dcf165e25c81b911f5ae4c0369e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 07:34:28.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f93a6682dd2f1121608a8df2eba198da62160015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f93a6682dd2f1121608a8df2eba198da62160015"}]},"branch":"refs/heads/master"},"c43713c287b7bee65be6df7ce15de53df6438f45":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-06 10:52:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/6"}}},"commit":{"parents":[{"commit":"63c08dbb228aa9e1d6cd835557a63c2d35e11d33","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/63c08dbb228aa9e1d6cd835557a63c2d35e11d33"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 10:51:43.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c43713c287b7bee65be6df7ce15de53df6438f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c43713c287b7bee65be6df7ce15de53df6438f45"}]},"branch":"refs/heads/master"},"3e2902a0db9a92a6cad42e31d89d5f0d114f9d40":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-06 12:18:56.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/7"}}},"commit":{"parents":[{"commit":"8c2962cbd128a8f18aff3647b6ffe989ccdb09ef","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8c2962cbd128a8f18aff3647b6ffe989ccdb09ef"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 12:14:57.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3e2902a0db9a92a6cad42e31d89d5f0d114f9d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3e2902a0db9a92a6cad42e31d89d5f0d114f9d40"}]},"branch":"refs/heads/master"},"8b0c01f72ba0c5187b07aa9b55e7263d14f32cf1":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-06 13:21:43.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/8"}}},"commit":{"parents":[{"commit":"1d49bf070b8323e1d5e6aab98d3544fa3f4a7230","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1d49bf070b8323e1d5e6aab98d3544fa3f4a7230"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 13:21:32.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b0c01f72ba0c5187b07aa9b55e7263d14f32cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b0c01f72ba0c5187b07aa9b55e7263d14f32cf1"}]},"branch":"refs/heads/master"},"8cda39490bb6f6ddda9f5f8415585895ea579918":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-07 08:19:03.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/9"}}},"commit":{"parents":[{"commit":"f943e9b4e62c65c9c482c66a145a15916a151ac2","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f943e9b4e62c65c9c482c66a145a15916a151ac2"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-07 08:17:08.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cda39490bb6f6ddda9f5f8415585895ea579918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cda39490bb6f6ddda9f5f8415585895ea579918"}]},"branch":"refs/heads/master"},"ae33dc307142edcbd4a31ec6f23014ac934d6913":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-09 07:25:35.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/89/672889/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/10"}}},"commit":{"parents":[{"commit":"9aa0c134d46798f7402f2dec783d9ed0d14c3a02","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9aa0c134d46798f7402f2dec783d9ed0d14c3a02"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-09 07:24:17.000000000","tz":120},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ae33dc307142edcbd4a31ec6f23014ac934d6913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ae33dc307142edcbd4a31ec6f23014ac934d6913"}]},"branch":"refs/heads/master"},"0cb6d6172ea078f4657594dc34c744a03c0420cd":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-08-22 17:57:01.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672889/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/89/672889/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/89/672889/11"}}},"commit":{"parents":[{"commit":"3ba8931a58d00c8f631ff8a32dd16377748a0fc8","subject":"Adds molecule testing for undercloud-heat-purge-deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3ba8931a58d00c8f631ff8a32dd16377748a0fc8"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-26 07:03:53.000000000","tz":120},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 17:57:01.000000000","tz":0},"subject":"Adds molecule testing for undercloud-tokenflush","message":"Adds molecule testing for undercloud-tokenflush\n\nChange-Id: I26275ce12bbb824a24a7d2571d1920208cd34aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0cb6d6172ea078f4657594dc34c744a03c0420cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0cb6d6172ea078f4657594dc34c744a03c0420cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
