)]}'
{"id":"openstack%2Ftripleo-validations~673007","triplet_id":"openstack%2Ftripleo-validations~master~I66b796fab1d8651163226febbc4e99648a9ecc6a","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework-infra","hashtags":[],"change_id":"I66b796fab1d8651163226febbc4e99648a9ecc6a","subject":"Linting hardening with pre-commit","status":"MERGED","created":"2019-07-26 15:28:52.000000000","updated":"2019-08-08 18:44:56.000000000","submitted":"2019-08-08 18:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673007-1565289896298-067e4e06","meta_rev_id":"7a58d456849c79631580f4283603d569276a0da0","_number":673007,"virtual_id_number":673007,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-08 09:48:31.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-08 18:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"date":"2019-08-07 11:27:16.000000000","_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":2,"date":"2019-08-06 20:33:53.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-08 09:58:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-07 05:22:50.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 17:55:43.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2019-07-30 22:07:32.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-31 14:17:38.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-06 20:33:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-07 05:22:50.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-07 11:27:16.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-08-08 09:58:42.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-08 18:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39f505cf8082144fcb958bc98341d40d298e1308","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-26 15:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1f4dba8fd483ffca91e8f4bd7be144a731118b7","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-26 15:33:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5028cd8decc6e31c2f09fa0ca8b0f3ef23829cd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:53:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-containers-multinode/e0b7efe/ : SUCCESS in 2h 27m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-undercloud-containers/7c65e19/ : SUCCESS in 1h 27m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e7ed57c/ : SUCCESS in 2h 20m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3815c47/ : SUCCESS in 2h 37m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-standalone/6d755e6/ : SUCCESS in 1h 51m 12s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/2/check/tripleo-ci-fedora-28-standalone/568c7a1/ : SUCCESS in 1h 42m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/2/check/tripleo-ci-centos-7-standalone-upgrade/f597d65/ : SUCCESS in 1h 58m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/2/check/openstack-tox-pep8/9cd98b8/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/2/check/openstack-tox-py27/96dd932/ : POST_FAILURE in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/2/check/openstack-tox-py36/f5b8767/ : POST_FAILURE in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/2/check/openstack-tox-py37/5b25c8d/ : POST_FAILURE in 5m 09s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/2/check/openstack-tox-molecule/6fe8c20/tox/reports.html : SUCCESS in 26m 30s\n- requirements-check http://logs.openstack.org/07/673007/2/check/requirements-check/1704d72/ : SUCCESS in 2m 23s\n- openstack-tox-docs http://logs.openstack.org/07/673007/2/check/openstack-tox-docs/f4a538e/ : POST_FAILURE in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/2/check/build-openstack-releasenotes/ea44946/html/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/2/check/openstack-tox-lower-constraints/a980c8a/ : FAILURE in 2m 07s","accounts_in_message":[],"_revision_number":2},{"id":"349823dd825560ee20df9357a3af07625478c84b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-28 13:35:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6448feabcef56aa0f28f4157b60c748e22e1cf4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-28 14:09:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20e084940ffd25c00151e781363451b853b2dc34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 16:54:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-containers-multinode/90e274d/ : SUCCESS in 2h 15m 44s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-undercloud-containers/78edddb/ : SUCCESS in 1h 25m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0417ec2/ : SUCCESS in 2h 06m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1821631/ : SUCCESS in 2h 38m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-standalone/153bcfc/ : SUCCESS in 1h 51m 14s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/4/check/tripleo-ci-fedora-28-standalone/974feb1/ : FAILURE in 1h 29m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/4/check/tripleo-ci-centos-7-standalone-upgrade/0624e30/ : SUCCESS in 2h 12m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/4/check/openstack-tox-pep8/f9493bc/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/4/check/openstack-tox-py27/97cb179/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/4/check/openstack-tox-py36/1f1f316/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/4/check/openstack-tox-py37/4579c52/ : SUCCESS in 3m 53s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/4/check/openstack-tox-molecule/e6f3896/tox/reports.html : SUCCESS in 24m 46s\n- requirements-check http://logs.openstack.org/07/673007/4/check/requirements-check/304497d/ : FAILURE in 2m 25s\n- openstack-tox-docs http://logs.openstack.org/07/673007/4/check/openstack-tox-docs/89b5870/html/ : SUCCESS in 4m 05s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/4/check/build-openstack-releasenotes/ad812aa/html/ : SUCCESS in 3m 44s\n- openstack-tox-linters http://logs.openstack.org/07/673007/4/check/openstack-tox-linters/e30454e/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/4/check/openstack-tox-lower-constraints/5a6988c/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"50015398a143f2d42e51f4633c4a298a1f4b2a87","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 10:24:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"531492cc4aa533e1d3d64ef2d761aad20ca1ceb5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 10:39:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"439bf598c45ec35456d5a2e14b1b98958a6c2721","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:18:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8992509640428d865dff4e9c27864f1d33d91e0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 11:19:14.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":7},{"id":"adc096bb63bc49adff98d5f098d5d256e14f45b9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:21:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d8e7a713f1ad9b4f2331c9a47b3988f9487b9277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 11:22:00.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":8},{"id":"1cd17bb20ace6ef056a128d7d0391a29aefae965","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:23:16.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b4ceef48522f705bee4e74718bd63976cb1fe037","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 11:36:54.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"a5edfc21b7da98da7741a164f2407555648b38b2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 12:27:39.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"f4e306ea466ec80ee664f9bf4f339b7f330bed08","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 13:07:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ecb396afa334d9c84315381a0731a799673b288f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 13:09:15.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d51bf3ea2542a5e2d511974c2f87374527466f30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 14:56:19.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-containers-multinode/de84a48/ : FAILURE in 28m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-undercloud-containers/eceef28/ : FAILURE in 25m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/32dad32/ : FAILURE in 28m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/04ea7b2/ : FAILURE in 28m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-standalone/ed629e3/ : FAILURE in 26m 11s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/13/check/tripleo-ci-fedora-28-standalone/095ed85/ : FAILURE in 26m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-standalone-upgrade/e1167f7/ : FAILURE in 1h 25m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/13/check/openstack-tox-pep8/c86042b/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/13/check/openstack-tox-py27/ab378d9/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/13/check/openstack-tox-py36/5b03fc4/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/13/check/openstack-tox-py37/66596ed/ : SUCCESS in 6m 18s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/13/check/openstack-tox-molecule/a8080ef/tox/reports.html : SUCCESS in 26m 17s\n- requirements-check http://logs.openstack.org/07/673007/13/check/requirements-check/03bfed1/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/07/673007/13/check/openstack-tox-docs/ac027df/html/ : SUCCESS in 6m 09s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/13/check/build-openstack-releasenotes/f170736/html/ : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/13/check/openstack-tox-lower-constraints/c0c936b/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":13},{"id":"4f0790af6dc7db047bf547dbcd124173fbf67280","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-07-29 17:55:43.000000000","message":"Patch Set 13:\n\nrecheck - seems unrelated","accounts_in_message":[],"_revision_number":13},{"id":"f3088f597668dfd0b8e12e6da06af1e75041c13a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:49:07.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-containers-multinode/63519ad/ : FAILURE in 27m 26s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-undercloud-containers/26a2a43/ : FAILURE in 26m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c652d72/ : FAILURE in 29m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5cf400c/ : FAILURE in 30m 25s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-standalone/535043d/ : FAILURE in 24m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/13/check/tripleo-ci-fedora-28-standalone/08857b7/ : FAILURE in 27m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/13/check/tripleo-ci-centos-7-standalone-upgrade/5d3f500/ : FAILURE in 1h 39m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/13/check/openstack-tox-pep8/16296ef/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/13/check/openstack-tox-py27/f56dff4/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/13/check/openstack-tox-py36/d32d306/ : SUCCESS in 6m 05s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/13/check/openstack-tox-py37/a99cb15/ : SUCCESS in 5m 00s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/13/check/openstack-tox-molecule/a5eceb1/tox/reports.html : SUCCESS in 28m 38s\n- requirements-check http://logs.openstack.org/07/673007/13/check/requirements-check/df024a2/ : SUCCESS in 2m 27s\n- openstack-tox-docs http://logs.openstack.org/07/673007/13/check/openstack-tox-docs/89711e9/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/13/check/build-openstack-releasenotes/46a2566/html/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/13/check/openstack-tox-lower-constraints/c78a535/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":13},{"id":"34ad5d6a92f185cc1a2405ec525035ebc7d77f86","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 09:47:38.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"3f7099984dc3ffa3d275b37adf1b9c64911b9948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 11:57:28.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-containers-multinode/2c28a07/ : FAILURE in 25m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-undercloud-containers/ec274e4/ : FAILURE in 21m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b42d373/ : FAILURE in 25m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6a9a361/ : FAILURE in 30m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-standalone/9a4c1d8/ : FAILURE in 24m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/14/check/tripleo-ci-fedora-28-standalone/4092325/ : FAILURE in 28m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/14/check/tripleo-ci-centos-7-standalone-upgrade/7deec16/ : FAILURE in 1h 22m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/14/check/openstack-tox-pep8/49f73b6/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/14/check/openstack-tox-py27/2cdd278/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/14/check/openstack-tox-py36/9402a57/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/14/check/openstack-tox-py37/3bce127/ : SUCCESS in 5m 32s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/14/check/openstack-tox-molecule/74b1f11/tox/reports.html : TIMED_OUT in 31m 29s\n- requirements-check http://logs.openstack.org/07/673007/14/check/requirements-check/196a719/ : SUCCESS in 2m 44s\n- openstack-tox-docs http://logs.openstack.org/07/673007/14/check/openstack-tox-docs/3735a65/ : FAILURE in 4m 38s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/14/check/build-openstack-releasenotes/8c71651/html/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/14/check/openstack-tox-lower-constraints/e12bfb9/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":14},{"id":"f2b3da05f8c29ac53d7bf041df27ee7221697276","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 12:15:21.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"ea4884d85b004221b44ff874cdc6374b3507557b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-30 14:01:36.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"1b5cd5351c63e75cd4dd60d9eaca0045dd76c306","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:51:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-containers-multinode/b11883e/ : SUCCESS in 2h 22m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-undercloud-containers/e0587da/ : SUCCESS in 1h 40m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a6c5bea/ : SUCCESS in 2h 17m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ac0fc70/ : SUCCESS in 2h 30m 37s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-standalone/0d035b5/ : SUCCESS in 1h 45m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/16/check/tripleo-ci-fedora-28-standalone/4d3c504/ : SUCCESS in 1h 43m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/16/check/tripleo-ci-centos-7-standalone-upgrade/7301695/ : SUCCESS in 2h 09m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/16/check/openstack-tox-pep8/a4995c2/ : FAILURE in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/16/check/openstack-tox-py27/2efddc6/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/16/check/openstack-tox-py36/23584d8/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/16/check/openstack-tox-py37/5539bf6/ : SUCCESS in 4m 36s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/16/check/openstack-tox-molecule/f9471ee/tox/reports.html : SUCCESS in 25m 11s\n- requirements-check http://logs.openstack.org/07/673007/16/check/requirements-check/2ce5abb/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/07/673007/16/check/openstack-tox-docs/3deac74/ : FAILURE in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/16/check/build-openstack-releasenotes/b362ca8/html/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/16/check/openstack-tox-lower-constraints/9fadc70/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":16},{"id":"d29b9064644e62f07249d32c5d301744a396726a","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-30 22:07:32.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4e4d61c94ee32435ee6f8cdde5c53decabf2ad5a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 10:09:34.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a368f3c4fbd453faca2f40c26bc0703ba3843c36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 14:12:09.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-containers-multinode/c38f98c/ : SUCCESS in 2h 19m 35s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-undercloud-containers/c72ac02/ : SUCCESS in 1h 40m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/65ea9ce/ : SUCCESS in 2h 32m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/58d3ca4/ : SUCCESS in 2h 23m 40s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-standalone/d0d404c/ : SUCCESS in 1h 47m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/17/check/tripleo-ci-fedora-28-standalone/65a1866/ : FAILURE in 1h 48m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/17/check/tripleo-ci-centos-7-standalone-upgrade/05194b1/ : SUCCESS in 1h 56m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/17/check/openstack-tox-pep8/b090dc3/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/17/check/openstack-tox-py27/0645858/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/17/check/openstack-tox-py36/0d9ce5b/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/17/check/openstack-tox-py37/e541eed/ : SUCCESS in 5m 03s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/17/check/openstack-tox-molecule/6b55824/tox/reports.html : SUCCESS in 27m 37s\n- requirements-check http://logs.openstack.org/07/673007/17/check/requirements-check/a4e49dd/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/07/673007/17/check/openstack-tox-docs/fd26dc3/ : FAILURE in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/17/check/build-openstack-releasenotes/7e7c167/html/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/17/check/openstack-tox-lower-constraints/8150a44/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":17},{"id":"2dca7c4a718fe4fcaf270ea48019a9f5771776ec","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-31 14:17:38.000000000","message":"Patch Set 17: Code-Review-1\n\nLooks like a small issue, we just need to not include the no-op doc that does-not-exist (or is located elsewhere?): http://logs.openstack.org/07/673007/17/check/openstack-tox-docs/fd26dc3/job-output.txt.gz#_2019-07-31_11_45_46_705629","accounts_in_message":[],"_revision_number":17},{"id":"4489af1380226d690bbe7b8da7b8743f628c909b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 14:32:49.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f7ad2da2fbdb24070cac6009d4085737511545ba","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 14:57:37.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"641ce9e25b9ec3036e62dedf8ad147b304deeb7e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-31 15:48:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"68b378cfae14efb22fe1b7f41bd9b6e711b60870","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 21:23:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-containers-multinode/8c30f14/ : FAILURE in 34m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-undercloud-containers/d5a942c/ : FAILURE in 26m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/31af8e2/ : FAILURE in 32m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0832adc/ : FAILURE in 33m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-standalone/a9b62a9/ : FAILURE in 28m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/07/673007/20/check/tripleo-ci-fedora-28-standalone/adf7cce/ : FAILURE in 28m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/07/673007/20/check/tripleo-ci-centos-7-standalone-upgrade/5642def/ : FAILURE in 1h 22m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/673007/20/check/openstack-tox-pep8/89303ca/ : SUCCESS in 8m 42s\n- openstack-tox-py27 http://logs.openstack.org/07/673007/20/check/openstack-tox-py27/aaa5db7/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/07/673007/20/check/openstack-tox-py36/9260342/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/07/673007/20/check/openstack-tox-py37/dc76af4/ : SUCCESS in 6m 20s\n- openstack-tox-molecule http://logs.openstack.org/07/673007/20/check/openstack-tox-molecule/e0226b4/tox/reports.html : SUCCESS in 27m 09s\n- requirements-check http://logs.openstack.org/07/673007/20/check/requirements-check/56efda8/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/07/673007/20/check/openstack-tox-docs/4856ba9/html/ : SUCCESS in 6m 51s\n- build-openstack-releasenotes http://logs.openstack.org/07/673007/20/check/build-openstack-releasenotes/07fd7c3/html/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/673007/20/check/openstack-tox-lower-constraints/bbb73ee/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":20},{"id":"c59644ab23eac47382606fdeca7d3e9cb30f9bfc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 10:35:06.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"6484866b37dc8dc3defa0ae13ebf1f34aa2513d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:48:59.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-containers-multinode/fb42803/ : SUCCESS in 2h 04m 41s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-undercloud-containers/5c41e88/ : SUCCESS in 1h 31m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3869ddf/ : SUCCESS in 2h 08m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/80964bc/ : SUCCESS in 2h 09m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-standalone/5d688df/ : SUCCESS in 1h 34m 32s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/21/check/tripleo-ci-fedora-28-standalone/40c775d/ : FAILURE in 55m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/21/check/tripleo-ci-centos-7-standalone-upgrade/dfe792e/ : FAILURE in 1h 29m 27s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/21/check/openstack-tox-pep8/35d986b/ : SUCCESS in 4m 26s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/21/check/openstack-tox-py27/ec9a26c/ : SUCCESS in 4m 16s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/21/check/openstack-tox-py36/6749b94/ : SUCCESS in 3m 51s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/21/check/openstack-tox-py37/c151984/ : SUCCESS in 4m 12s\n- openstack-tox-molecule https://logs.opendev.org/07/673007/21/check/openstack-tox-molecule/5abb6d1/tox/reports.html : SUCCESS in 26m 24s\n- requirements-check https://logs.opendev.org/07/673007/21/check/requirements-check/1ff1f3a/ : SUCCESS in 2m 23s\n- openstack-tox-docs https://logs.opendev.org/07/673007/21/check/openstack-tox-docs/f20c6fd/html/ : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/21/check/build-openstack-releasenotes/2799397/html/ : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/21/check/openstack-tox-lower-constraints/498c517/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":21},{"id":"eaf41cf4b7b35d3de444328e12572de68815cbce","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 14:16:02.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"2f6db9d18c3c7b2bb88135eafd239505f18dae23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 17:32:04.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-containers-multinode/54febee/ : SUCCESS in 2h 07m 51s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-undercloud-containers/6a28ca3/ : SUCCESS in 1h 51m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e733c0e/ : SUCCESS in 2h 11m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7d2b1dd/ : SUCCESS in 2h 40m 39s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-standalone/4298090/ : SUCCESS in 1h 48m 02s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/22/check/tripleo-ci-fedora-28-standalone/35f8d43/ : FAILURE in 56m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/22/check/tripleo-ci-centos-7-standalone-upgrade/0dcf80a/ : FAILURE in 1h 42m 52s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/22/check/openstack-tox-pep8/6b2dfc8/ : SUCCESS in 4m 15s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/22/check/openstack-tox-py27/2c9b491/ : SUCCESS in 4m 47s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/22/check/openstack-tox-py36/07528b3/ : SUCCESS in 5m 35s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/22/check/openstack-tox-py37/d6ba8a6/ : SUCCESS in 4m 45s\n- openstack-tox-molecule https://logs.opendev.org/07/673007/22/check/openstack-tox-molecule/8d25b8f/tox/reports.html : SUCCESS in 26m 04s\n- requirements-check https://logs.opendev.org/07/673007/22/check/requirements-check/2800f4d/ : SUCCESS in 2m 22s\n- openstack-tox-docs https://logs.opendev.org/07/673007/22/check/openstack-tox-docs/bd3dc1a/html/ : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/22/check/build-openstack-releasenotes/f08cb1d/html/ : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/22/check/openstack-tox-lower-constraints/f08010c/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":22},{"id":"e7242e5e54d647fa53aa491db9f8c7ddd993e641","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-05 07:10:28.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"4b0807f62e7bfbf83a0c49a6ebbe3201e42e11fc","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 13:05:00.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"c01bcde6dfac99c49515557c8da284bfd98ca010","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 13:45:12.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"0a505b177c79eeff73ca027273e418b351bcaed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:45:25.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"230bc9688591013f79751fa89295a4e06c48e497","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:31:03.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/22/gate/tripleo-ci-centos-7-containers-multinode/cc28e6e/ : SUCCESS in 2h 32m 59s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/22/gate/tripleo-ci-centos-7-standalone/25016b7/ : SUCCESS in 1h 46m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/22/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5c1cc51/ : SUCCESS in 2h 24m 02s\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/22/gate/openstack-tox-pep8/5908faf/ : SUCCESS in 6m 43s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/22/gate/openstack-tox-py27/b8360ec/ : SUCCESS in 4m 54s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/22/gate/openstack-tox-py36/d8dac2a/ : SUCCESS in 4m 54s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/22/gate/openstack-tox-py37/3ca7464/ : SUCCESS in 5m 32s\n- openstack-tox-molecule https://logs.opendev.org/07/673007/22/gate/openstack-tox-molecule/ba28ed3/tox/reports.html : TIMED_OUT in 31m 37s\n- requirements-check https://logs.opendev.org/07/673007/22/gate/requirements-check/037c10c/ : SUCCESS in 3m 40s\n- openstack-tox-docs https://logs.opendev.org/07/673007/22/gate/openstack-tox-docs/c3ca6bf/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/22/gate/build-openstack-releasenotes/5f658e6/html/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/22/gate/openstack-tox-lower-constraints/1a1a055/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":22},{"id":"96bdd66177b6662b685a1f83fb62254f56783d6b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 07:35:04.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d7d972be25fbd3a7c7e9d255d64d8929284a9d45","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-06 09:12:20.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"dd5fe9c77976daa87e96fdeba293c4c6523de0ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 10:42:58.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-containers-multinode/4065404/ : SUCCESS in 2h 31m 30s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-undercloud-containers/5c3aa2d/ : SUCCESS in 1h 41m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9b5e782/ : SUCCESS in 2h 14m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3a33432/ : SUCCESS in 2h 31m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-standalone/524af25/ : SUCCESS in 1h 52m 58s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/23/check/tripleo-ci-fedora-28-standalone/42c49e4/ : SUCCESS in 1h 46m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/23/check/tripleo-ci-centos-7-standalone-upgrade/71e8906/ : SUCCESS in 2h 08m 43s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/23/check/openstack-tox-pep8/0673848/ : SUCCESS in 6m 04s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/23/check/openstack-tox-py27/1136f12/ : SUCCESS in 5m 18s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/23/check/openstack-tox-py36/6498ca7/ : SUCCESS in 5m 24s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/23/check/openstack-tox-py37/17a87ac/ : SUCCESS in 4m 30s\n- requirements-check finger://ze02.openstack.org/2a29786ca3d3425593f5278102b65aff : POST_FAILURE in 3m 04s\n- openstack-tox-docs https://logs.opendev.org/07/673007/23/check/openstack-tox-docs/4eb6fde/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/23/check/build-openstack-releasenotes/c280b4f/html/ : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/23/check/openstack-tox-lower-constraints/8805ee2/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":23},{"id":"da410e40275c760f869e89f06802ebdbd2fce5e8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 10:52:34.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"9a52b2b4cc330da01cec8f2669d088c6f8f8bf05","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 12:18:56.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"28269f1093eaa39aa38bd73dd1dd42bb50932168","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 13:21:43.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"b5fb5dee1d51dad1ff0f18812f4edae513c5aa3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:41:54.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-containers-multinode/49fca6f/ : SUCCESS in 2h 40m 37s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-undercloud-containers/a1a0a2a/ : SUCCESS in 1h 59m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/04a8e05/ : SUCCESS in 2h 11m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d8422f5/ : SUCCESS in 2h 17m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone/f8524f0/ : SUCCESS in 1h 47m 00s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-fedora-28-standalone/6ef1521/ : SUCCESS in 1h 46m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone-upgrade/9b927d1/ : SUCCESS in 2h 07m 06s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/26/check/openstack-tox-pep8/e81c221/ : SUCCESS in 4m 42s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/26/check/openstack-tox-py27/756ff99/ : SUCCESS in 4m 18s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/26/check/openstack-tox-py36/b0e544d/ : SUCCESS in 8m 11s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/26/check/openstack-tox-py37/b665181/ : SUCCESS in 9m 46s\n- requirements-check https://logs.opendev.org/07/673007/26/check/requirements-check/393f2ac/ : SUCCESS in 3m 08s\n- openstack-tox-docs https://logs.opendev.org/07/673007/26/check/openstack-tox-docs/1aa3e86/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/26/check/build-openstack-releasenotes/d7955ff/html/ : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/26/check/openstack-tox-lower-constraints/83014ac/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":26},{"id":"27b68732e1363cada36ab0927ff14c15c8236e19","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 20:33:53.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"cbcde7db64118415823e917e006dadb165d31995","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-07 05:22:50.000000000","message":"Patch Set 26: Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"851eb99de81612f413f7df58df907bf42010c6b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 05:23:04.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"d6c116675b98a0fff77b03e77b83b184bd56680c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 08:56:57.000000000","message":"Patch Set 26: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-containers-multinode/da5a32e/ : SUCCESS in 2h 33m 22s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-standalone/db01773/ : TIMED_OUT in 3h 24m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5c75b31/ : SUCCESS in 2h 48m 57s\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/26/gate/openstack-tox-pep8/5db1303/ : SUCCESS in 4m 03s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py27/44897c8/ : SUCCESS in 4m 02s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py36/8f63c02/ : SUCCESS in 4m 27s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py37/bbd49a4/ : SUCCESS in 5m 55s\n- requirements-check https://logs.opendev.org/07/673007/26/gate/requirements-check/7f49ef6/ : SUCCESS in 2m 41s\n- openstack-tox-docs https://logs.opendev.org/07/673007/26/gate/openstack-tox-docs/1adb32a/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/26/gate/build-openstack-releasenotes/0e61d72/html/ : SUCCESS in 2m 56s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/26/gate/openstack-tox-lower-constraints/73c91f3/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":26},{"id":"76b2cfc213562ce4eadf2f70478ea736f1bae1db","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-08-07 11:27:16.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"a538d3bf34b74bb445e4d420d2b9630e5bcc9215","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 16:21:49.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-containers-multinode/9063b7d/ : FAILURE in 1h 11m 47s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-undercloud-containers/48ef347/ : FAILURE in 1h 06m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a96ba01/ : SUCCESS in 2h 42m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f9fbfd2/ : SUCCESS in 2h 30m 23s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone/a09bf56/ : SUCCESS in 2h 13m 43s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-fedora-28-standalone/6c13ba4/ : SUCCESS in 2h 30m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone-upgrade/d530543/ : SUCCESS in 2h 11m 42s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/26/check/openstack-tox-pep8/99b7293/ : SUCCESS in 4m 43s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/26/check/openstack-tox-py27/c0f8851/ : SUCCESS in 5m 13s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/26/check/openstack-tox-py36/cab5643/ : SUCCESS in 5m 51s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/26/check/openstack-tox-py37/5dd02a1/ : SUCCESS in 6m 46s\n- requirements-check https://logs.opendev.org/07/673007/26/check/requirements-check/67ea3c8/ : SUCCESS in 3m 18s\n- openstack-tox-docs https://logs.opendev.org/07/673007/26/check/openstack-tox-docs/99bea13/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/26/check/build-openstack-releasenotes/92e54bc/html/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/26/check/openstack-tox-lower-constraints/57a196d/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":26},{"id":"5ad91811e1e49a4967aa45d77323eb1802bec213","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-08 09:48:31.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"9b3a02d617a97ac36c555df7371b395727bda453","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-08 09:58:42.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"461c0d837144d116744592d6e5d153828928d9ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:39:59.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-containers-multinode/aa491e0/ : SUCCESS in 2h 31m 42s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-undercloud-containers/e0a550c/ : SUCCESS in 1h 48m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/897317e/ : SUCCESS in 2h 27m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5a80f3c/ : TIMED_OUT in 3h 23m 21s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone/983e5e7/ : SUCCESS in 1h 51m 18s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/07/673007/26/check/tripleo-ci-fedora-28-standalone/ffbc3e5/ : FAILURE in 2h 05m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/07/673007/26/check/tripleo-ci-centos-7-standalone-upgrade/ebbba6c/ : SUCCESS in 2h 18m 26s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/26/check/openstack-tox-pep8/a146002/ : SUCCESS in 3m 47s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/26/check/openstack-tox-py27/17a1bb8/ : SUCCESS in 4m 56s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/26/check/openstack-tox-py36/b7ef8f8/ : SUCCESS in 5m 47s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/26/check/openstack-tox-py37/4ce3949/ : SUCCESS in 4m 11s\n- requirements-check https://logs.opendev.org/07/673007/26/check/requirements-check/70298fa/ : SUCCESS in 2m 28s\n- openstack-tox-docs https://logs.opendev.org/07/673007/26/check/openstack-tox-docs/5944f29/html/ : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/26/check/build-openstack-releasenotes/37530dd/html/ : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/26/check/openstack-tox-lower-constraints/c6da75a/ : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":26},{"id":"20698b6964d03b4ac1bb670ccb8edccf0d071334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:40:25.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"4d9567878f0db809eb16ebc9d60a843ba5487a61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:44:56.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-containers-multinode/7f2dbaa/ : SUCCESS in 3h 05m 12s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-standalone/9803fcb/ : SUCCESS in 1h 50m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/07/673007/26/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9f954c9/ : SUCCESS in 2h 26m 47s\n- openstack-tox-pep8 https://logs.opendev.org/07/673007/26/gate/openstack-tox-pep8/f13761a/ : SUCCESS in 6m 27s\n- openstack-tox-py27 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py27/f199ad3/ : SUCCESS in 4m 29s\n- openstack-tox-py36 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py36/a7fe5f4/ : SUCCESS in 4m 26s\n- openstack-tox-py37 https://logs.opendev.org/07/673007/26/gate/openstack-tox-py37/dea0a28/ : SUCCESS in 6m 06s\n- requirements-check https://logs.opendev.org/07/673007/26/gate/requirements-check/8194d0c/ : SUCCESS in 3m 02s\n- openstack-tox-docs https://logs.opendev.org/07/673007/26/gate/openstack-tox-docs/18f93e1/html/ : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://logs.opendev.org/07/673007/26/gate/build-openstack-releasenotes/35d59ec/html/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/07/673007/26/gate/openstack-tox-lower-constraints/6b6b4bf/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":26},{"id":"957146782947b4cc30850554041a2f4571a60cd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:44:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"cac3e7101d89c743ce2593141b9da4735be2fd33","revisions":{"1a71ea8772b032d691bb46e3b9737fc745fccb8e":{"kind":"REWORK","_number":1,"created":"2019-07-26 15:28:52.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/1"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:28:37.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1a71ea8772b032d691bb46e3b9737fc745fccb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1a71ea8772b032d691bb46e3b9737fc745fccb8e"}]},"branch":"refs/heads/master"},"e6adfb66ec58fb7f76ecd20b70014a2abc7d016d":{"kind":"REWORK","_number":2,"created":"2019-07-26 15:33:44.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/2"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:32:41.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e6adfb66ec58fb7f76ecd20b70014a2abc7d016d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e6adfb66ec58fb7f76ecd20b70014a2abc7d016d"}]},"branch":"refs/heads/master"},"364dab87298bc7bd52bc3933fbf9da3adf40a328":{"kind":"REWORK","_number":3,"created":"2019-07-28 13:35:53.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/3"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-28 13:35:44.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/364dab87298bc7bd52bc3933fbf9da3adf40a328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/364dab87298bc7bd52bc3933fbf9da3adf40a328"}]},"branch":"refs/heads/master"},"6906ab9acda2ed525b3d59e4fde913cb9d6b7ed8":{"kind":"REWORK","_number":4,"created":"2019-07-28 14:09:44.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/4"}}},"commit":{"parents":[{"commit":"fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e","subject":"Correct \"repos\" validation and adds Molecule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa974b9fb9f1e9d3d2ebe5ec5d95095ea018703e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-28 14:09:23.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6906ab9acda2ed525b3d59e4fde913cb9d6b7ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6906ab9acda2ed525b3d59e4fde913cb9d6b7ed8"}]},"branch":"refs/heads/master"},"b4ef818e8e1677eee4cc1fbf8d5e57dd2ce43cc3":{"kind":"REWORK","_number":5,"created":"2019-07-29 10:24:57.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/5"}}},"commit":{"parents":[{"commit":"0851d91af936fd3b56459dab1293afbd8547379f","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0851d91af936fd3b56459dab1293afbd8547379f"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 10:24:13.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b4ef818e8e1677eee4cc1fbf8d5e57dd2ce43cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b4ef818e8e1677eee4cc1fbf8d5e57dd2ce43cc3"}]},"branch":"refs/heads/master"},"9990a44a6b37a6ea558c87ef0afa7d7659eaeadb":{"kind":"REWORK","_number":6,"created":"2019-07-29 10:39:44.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/6"}}},"commit":{"parents":[{"commit":"0851d91af936fd3b56459dab1293afbd8547379f","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0851d91af936fd3b56459dab1293afbd8547379f"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-26 15:08:20.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 10:39:32.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9990a44a6b37a6ea558c87ef0afa7d7659eaeadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9990a44a6b37a6ea558c87ef0afa7d7659eaeadb"}]},"branch":"refs/heads/master"},"6adf9d4b43375d51aa1749f152d1d74bdba587d8":{"kind":"REWORK","_number":7,"created":"2019-07-29 11:18:54.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/7"}}},"commit":{"parents":[{"commit":"323e2f77879f11b88ac5e661da178778b845f743","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/323e2f77879f11b88ac5e661da178778b845f743"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6adf9d4b43375d51aa1749f152d1d74bdba587d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6adf9d4b43375d51aa1749f152d1d74bdba587d8"}]},"branch":"refs/heads/master"},"c76bd3ae08a73784febb52a2ee3e9c3cdb14fc34":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-29 11:21:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/8"}}},"commit":{"parents":[{"commit":"c4a32be1e98eea90e866dd56443da626ec439591","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c4a32be1e98eea90e866dd56443da626ec439591"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:21:24.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c76bd3ae08a73784febb52a2ee3e9c3cdb14fc34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c76bd3ae08a73784febb52a2ee3e9c3cdb14fc34"}]},"branch":"refs/heads/master"},"83f1af18edcecdeca9b55b6351e04495653b357f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-07-29 11:23:16.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/9"}}},"commit":{"parents":[{"commit":"4b8ee8f5baab2f79c51f2e4fd5aafd7c22f23fde","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4b8ee8f5baab2f79c51f2e4fd5aafd7c22f23fde"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:23:07.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/83f1af18edcecdeca9b55b6351e04495653b357f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/83f1af18edcecdeca9b55b6351e04495653b357f"}]},"branch":"refs/heads/master"},"4d0faf1758d3cff37aad9189f708598645e4f749":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-07-29 11:36:54.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/10"}}},"commit":{"parents":[{"commit":"d8850e0c4aa84f70f04cf729d16685564012c305","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8850e0c4aa84f70f04cf729d16685564012c305"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:35:35.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4d0faf1758d3cff37aad9189f708598645e4f749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4d0faf1758d3cff37aad9189f708598645e4f749"}]},"branch":"refs/heads/master"},"93dc986bfde50882f540149ebe9f93ff60886c0f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-07-29 12:27:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/11"}}},"commit":{"parents":[{"commit":"7abcd6e08e9b171fb9eded6433703a1423d1f500","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7abcd6e08e9b171fb9eded6433703a1423d1f500"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 12:24:39.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/93dc986bfde50882f540149ebe9f93ff60886c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/93dc986bfde50882f540149ebe9f93ff60886c0f"}]},"branch":"refs/heads/master"},"11253738201b9da889274ba6348886c4372a0c97":{"kind":"REWORK","_number":12,"created":"2019-07-29 13:07:23.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/12"}}},"commit":{"parents":[{"commit":"7abcd6e08e9b171fb9eded6433703a1423d1f500","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7abcd6e08e9b171fb9eded6433703a1423d1f500"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 12:59:24.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/11253738201b9da889274ba6348886c4372a0c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/11253738201b9da889274ba6348886c4372a0c97"}]},"branch":"refs/heads/master"},"f7fa05a2d664ab4f22a778074151a07cdcf9b4cb":{"kind":"REWORK","_number":13,"created":"2019-07-29 13:09:15.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/13"}}},"commit":{"parents":[{"commit":"7abcd6e08e9b171fb9eded6433703a1423d1f500","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7abcd6e08e9b171fb9eded6433703a1423d1f500"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 11:18:09.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 13:08:51.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f7fa05a2d664ab4f22a778074151a07cdcf9b4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f7fa05a2d664ab4f22a778074151a07cdcf9b4cb"}]},"branch":"refs/heads/master"},"623e63ada9816d3c66d96f21e010312faf524b2c":{"kind":"REWORK","_number":14,"created":"2019-07-30 09:47:38.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/14"}}},"commit":{"parents":[{"commit":"876da38327115135d2deccf965e1e08d29f124f7","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/876da38327115135d2deccf965e1e08d29f124f7"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/623e63ada9816d3c66d96f21e010312faf524b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/623e63ada9816d3c66d96f21e010312faf524b2c"}]},"branch":"refs/heads/master"},"83267a17de5d9730f375f6af747a4d91b4073ff7":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-07-30 12:15:21.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/15"}}},"commit":{"parents":[{"commit":"9c203f81fe270ea9a9165940268f4320a130bf3c","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9c203f81fe270ea9a9165940268f4320a130bf3c"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 11:54:04.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/83267a17de5d9730f375f6af747a4d91b4073ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/83267a17de5d9730f375f6af747a4d91b4073ff7"}]},"branch":"refs/heads/master"},"f40f43904239332add62cd45d86e6bdc35c4a62e":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-07-30 14:01:36.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/16"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 13:54:29.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f40f43904239332add62cd45d86e6bdc35c4a62e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f40f43904239332add62cd45d86e6bdc35c4a62e"}]},"branch":"refs/heads/master"},"49eaf9276b9940f44e2751c82e30bb16ab907979":{"kind":"REWORK","_number":17,"created":"2019-07-31 10:09:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/17"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 10:09:20.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49eaf9276b9940f44e2751c82e30bb16ab907979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49eaf9276b9940f44e2751c82e30bb16ab907979"}]},"branch":"refs/heads/master"},"4e14d5895f11e8b018ec718aa240f8fc88b1854c":{"kind":"REWORK","_number":18,"created":"2019-07-31 14:32:49.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/18"}}},"commit":{"parents":[{"commit":"d3fc9eeede8860628c7693ee6626c49d4fd275d3","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3fc9eeede8860628c7693ee6626c49d4fd275d3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 14:32:30.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4e14d5895f11e8b018ec718aa240f8fc88b1854c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4e14d5895f11e8b018ec718aa240f8fc88b1854c"}]},"branch":"refs/heads/master"},"45e97e29d053c30f6384e5c68b2021170b8a9a64":{"kind":"REWORK","_number":19,"created":"2019-07-31 14:57:37.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/19"}}},"commit":{"parents":[{"commit":"22a78e6afbe84e30d0d4ab9e519edf12b55a5998","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/22a78e6afbe84e30d0d4ab9e519edf12b55a5998"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 14:54:02.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/45e97e29d053c30f6384e5c68b2021170b8a9a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/45e97e29d053c30f6384e5c68b2021170b8a9a64"}]},"branch":"refs/heads/master"},"764c94033d79a13d5777d4047e55ee69df47e39c":{"kind":"REWORK","_number":20,"created":"2019-07-31 15:48:45.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/20"}}},"commit":{"parents":[{"commit":"22a78e6afbe84e30d0d4ab9e519edf12b55a5998","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/22a78e6afbe84e30d0d4ab9e519edf12b55a5998"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-31 15:48:13.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/764c94033d79a13d5777d4047e55ee69df47e39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/764c94033d79a13d5777d4047e55ee69df47e39c"}]},"branch":"refs/heads/master"},"d8eeb370130cc78a0b92ab1c68def79f5a4e2981":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-08-02 10:35:06.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/21"}}},"commit":{"parents":[{"commit":"5f53b804f5752904e96cc4ad39613a119277f8cc","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5f53b804f5752904e96cc4ad39613a119277f8cc"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-02 10:34:54.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8eeb370130cc78a0b92ab1c68def79f5a4e2981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8eeb370130cc78a0b92ab1c68def79f5a4e2981"}]},"branch":"refs/heads/master"},"2763412cbf048f2df9f821bbca84dd96e1affbd4":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-08-02 14:16:02.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/22"}}},"commit":{"parents":[{"commit":"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-02 14:11:03.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2763412cbf048f2df9f821bbca84dd96e1affbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2763412cbf048f2df9f821bbca84dd96e1affbd4"}]},"branch":"refs/heads/master"},"b3b7a6e85ed993271b868857c885e034f784e2ca":{"kind":"REWORK","_number":23,"created":"2019-08-06 07:35:04.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/23"}}},"commit":{"parents":[{"commit":"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 07:32:57.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b3b7a6e85ed993271b868857c885e034f784e2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b3b7a6e85ed993271b868857c885e034f784e2ca"}]},"branch":"refs/heads/master"},"cb3e83e1081024fb7f3814c09eb6824e7701eb2f":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-08-06 10:52:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/24"}}},"commit":{"parents":[{"commit":"dcf051ec42071974ba75276c20366f57d4d045e9","subject":"Merge \"Improve Ceph health checks to work with custom cluster names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/dcf051ec42071974ba75276c20366f57d4d045e9"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 10:22:40.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cb3e83e1081024fb7f3814c09eb6824e7701eb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cb3e83e1081024fb7f3814c09eb6824e7701eb2f"}]},"branch":"refs/heads/master"},"5c538c5b7be9f78bae754df3e4d44d0af0b26786":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2019-08-06 12:18:56.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/25"}}},"commit":{"parents":[{"commit":"1e8a20510d534cd94a2334d42308044e8818cb0a","subject":"Merge \"Simplifying the check of the latest package version validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1e8a20510d534cd94a2334d42308044e8818cb0a"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 12:14:57.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5c538c5b7be9f78bae754df3e4d44d0af0b26786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5c538c5b7be9f78bae754df3e4d44d0af0b26786"}]},"branch":"refs/heads/master"},"cac3e7101d89c743ce2593141b9da4735be2fd33":{"kind":"REWORK","_number":26,"created":"2019-08-06 13:21:43.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/07/673007/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/07/673007/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/07/673007/26"}}},"commit":{"parents":[{"commit":"1e8a20510d534cd94a2334d42308044e8818cb0a","subject":"Merge \"Simplifying the check of the latest package version validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1e8a20510d534cd94a2334d42308044e8818cb0a"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-06 13:21:25.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cac3e7101d89c743ce2593141b9da4735be2fd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cac3e7101d89c743ce2593141b9da4735be2fd33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
