)]}'
{"id":"openstack%2Ftripleo-validations~673994","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"optimization/no-fact-gathering-stable/stein","hashtags":[],"change_id":"Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed","subject":"Enable fact gathering for needed validations","status":"MERGED","created":"2019-08-01 10:34:06.000000000","updated":"2019-08-06 15:48:08.000000000","submitted":"2019-08-06 15:48:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673994-1565106488976-72314eed","meta_rev_id":"586cf62c4f87f3f2000258aa0fe00af429f4adb6","_number":673994,"virtual_id_number":673994,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-02 09:06:32.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-06 15:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-06 05:41:44.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-03 11:54:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-08-06 13:02:50.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 11:54:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-06 05:41:44.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-06 13:02:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-06 15:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6010da214fc79de1c4dd4186184599ccb6bb6a3","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 10:34:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bae87df303a0bd7e23edf914e13b67c49c7b6c66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 14:12:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-containers-multinode/3e99aab/ : SUCCESS in 2h 16m 05s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-undercloud-containers/7c1cd6a/ : SUCCESS in 1h 41m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1530c55/ : TIMED_OUT in 3h 25m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f338738/ : SUCCESS in 2h 35m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-standalone/fe35005/ : SUCCESS in 2h 05m 27s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/94/673994/1/check/tripleo-ci-fedora-28-standalone/2c22266/ : SUCCESS in 1h 47m 12s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/94/673994/1/check/openstack-tox-pep8/9c93965/ : SUCCESS in 4m 20s\n- openstack-tox-py27 https://logs.opendev.org/94/673994/1/check/openstack-tox-py27/9ea1ac6/ : SUCCESS in 3m 44s\n- openstack-tox-py35 https://logs.opendev.org/94/673994/1/check/openstack-tox-py35/014fcec/ : SUCCESS in 7m 34s\n- openstack-tox-py36 https://logs.opendev.org/94/673994/1/check/openstack-tox-py36/1f2690d/ : SUCCESS in 4m 05s\n- openstack-tox-docs https://logs.opendev.org/94/673994/1/check/openstack-tox-docs/9c1190a/html/ : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/94/673994/1/check/openstack-tox-lower-constraints/9a5ea4b/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5e661f3df7c40a23ab35ca189610a3ba35f8183f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 09:06:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62fd579194beb7558c24811fa854b08cdc611e7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:00:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-containers-multinode/e2fc0d8/ : SUCCESS in 1h 58m 45s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-undercloud-containers/a2053c9/ : SUCCESS in 1h 57m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/352a463/ : SUCCESS in 2h 36m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8b03ca6/ : SUCCESS in 2h 39m 50s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/94/673994/1/check/tripleo-ci-centos-7-standalone/2ae2d29/ : SUCCESS in 1h 57m 32s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/94/673994/1/check/tripleo-ci-fedora-28-standalone/c88b3f5/ : SUCCESS in 1h 49m 56s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/94/673994/1/check/openstack-tox-pep8/216a20d/ : SUCCESS in 5m 13s\n- openstack-tox-py27 https://logs.opendev.org/94/673994/1/check/openstack-tox-py27/3ee6835/ : SUCCESS in 4m 15s\n- openstack-tox-py35 https://logs.opendev.org/94/673994/1/check/openstack-tox-py35/9de6454/ : SUCCESS in 4m 17s\n- openstack-tox-py36 https://logs.opendev.org/94/673994/1/check/openstack-tox-py36/e64e21f/ : SUCCESS in 5m 08s\n- openstack-tox-molecule https://logs.opendev.org/94/673994/1/check/openstack-tox-molecule/d4ee022/tox/reports.html : SUCCESS in 18m 49s\n- openstack-tox-docs https://logs.opendev.org/94/673994/1/check/openstack-tox-docs/ec3d910/html/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://logs.opendev.org/94/673994/1/check/openstack-tox-lower-constraints/b7f71d0/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b1fd818f6af70f9b50cc5eb3cc9c5f8aa7060e61","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-06 05:41:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fb6aefe5db7bad689f137b8ebbf531ef560c1e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 13:02:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae9bce6a8f48e5c4489f1acdc1d49b971733478d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:03:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c4a7860e782be23d4b57c40a06bf8de558c886b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:48:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/94/673994/1/gate/tripleo-ci-centos-7-containers-multinode/5a2998c/ : SUCCESS in 2h 07m 26s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/94/673994/1/gate/tripleo-ci-centos-7-standalone/0c33147/ : SUCCESS in 1h 48m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/94/673994/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/00c7a3e/ : SUCCESS in 2h 24m 41s\n- openstack-tox-pep8 https://logs.opendev.org/94/673994/1/gate/openstack-tox-pep8/8c3505f/ : SUCCESS in 6m 23s\n- openstack-tox-py27 https://logs.opendev.org/94/673994/1/gate/openstack-tox-py27/8e5c4fa/ : SUCCESS in 7m 11s\n- openstack-tox-py35 https://logs.opendev.org/94/673994/1/gate/openstack-tox-py35/c6d873b/ : SUCCESS in 5m 56s\n- openstack-tox-py36 https://logs.opendev.org/94/673994/1/gate/openstack-tox-py36/190d51e/ : SUCCESS in 4m 55s\n- openstack-tox-molecule https://logs.opendev.org/94/673994/1/gate/openstack-tox-molecule/d22325f/tox/reports.html : SUCCESS in 21m 43s\n- openstack-tox-docs https://logs.opendev.org/94/673994/1/gate/openstack-tox-docs/a405d23/html/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://logs.opendev.org/94/673994/1/gate/openstack-tox-lower-constraints/150f1f8/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"af884b659efde198f1202fea1da490344880813b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:48:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"000718fa93aced3beea549ee0c810572a1836168","revisions":{"000718fa93aced3beea549ee0c810572a1836168":{"kind":"REWORK","_number":1,"created":"2019-08-01 10:34:06.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/94/673994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/94/673994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/94/673994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/94/673994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/94/673994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/94/673994/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-18 11:25:08.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 10:34:06.000000000","tz":0},"subject":"Enable fact gathering for needed validations","message":"Enable fact gathering for needed validations\n\nImplicit fact gathering is disabled since\nhttps://review.opendev.org/672278\n\nThis means, for some validations, we must explicitly enable fact\ngathering.\n\nDepends-On: https://review.opendev.org/672278\nChange-Id: Iaf4e8f3f1bf31293f18aedf2d0ab253c04d83bed\n(cherry picked from commit 2affbb11f82afcdfcd77c282f718c92595cad291)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/000718fa93aced3beea549ee0c810572a1836168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/000718fa93aced3beea549ee0c810572a1836168"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
