)]}'
{"id":"openstack%2Ftripleo-validations~673995","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"validate/containers-stable/stein","hashtags":[],"change_id":"Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf","subject":"New validation: ensure services status","status":"MERGED","created":"2019-08-01 11:01:17.000000000","updated":"2019-08-01 19:42:53.000000000","submitted":"2019-08-01 19:42:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673995-1564688573959-e5490da1","meta_rev_id":"4f710a680f9cacd8ca1b94af033958018aa424bf","_number":673995,"virtual_id_number":673995,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-01 19:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 13:34:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-01 11:01:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-01 13:43:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 11:01:17.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:01:17.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:01:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 13:43:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-01 19:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26ce7e9d8cfabf9f887559b4e96871c063593fd2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:01:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"07b2c1868500865450ac4e6fbf96690315c98c64","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 13:34:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1a0adf8e7ccecabe154d48ab8c4900d82cbde22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 13:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/95/673995/1/check/tripleo-ci-centos-7-containers-multinode/b59cfa2/ : SUCCESS in 2h 06m 27s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/95/673995/1/check/tripleo-ci-centos-7-undercloud-containers/27327c2/ : SUCCESS in 2h 00m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/95/673995/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/06ecc02/ : SUCCESS in 2h 09m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/95/673995/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/279c758/ : SUCCESS in 2h 31m 36s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/95/673995/1/check/tripleo-ci-centos-7-standalone/b87cfe7/ : SUCCESS in 2h 38m 42s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/95/673995/1/check/tripleo-ci-fedora-28-standalone/4085b0e/ : SUCCESS in 1h 51m 31s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/95/673995/1/check/openstack-tox-pep8/d5070b0/ : SUCCESS in 4m 58s\n- openstack-tox-py27 https://logs.opendev.org/95/673995/1/check/openstack-tox-py27/93975b7/ : SUCCESS in 4m 01s\n- openstack-tox-py35 https://logs.opendev.org/95/673995/1/check/openstack-tox-py35/a74c520/ : SUCCESS in 5m 04s\n- openstack-tox-py36 https://logs.opendev.org/95/673995/1/check/openstack-tox-py36/34731d8/ : SUCCESS in 5m 07s\n- openstack-tox-docs https://logs.opendev.org/95/673995/1/check/openstack-tox-docs/6f10612/html/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/95/673995/1/check/openstack-tox-lower-constraints/4088a36/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e76f22df828f4810ff49a445e7dee977c179d0cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 13:43:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31b0833fe0c732fb79858c2c5d217f551c25c568","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 13:44:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a596402f3ebd3d6321841061cd14487ad57b37b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 15:02:43.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b410f1b46c9b9edb6bf060af9c96c58be7ffb9a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:42:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/95/673995/1/gate/tripleo-ci-centos-7-containers-multinode/d9d01a6/ : SUCCESS in 2h 15m 54s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/95/673995/1/gate/tripleo-ci-centos-7-standalone/2b2cd8c/ : SUCCESS in 1h 55m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/95/673995/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a546449/ : SUCCESS in 2h 54m 15s\n- openstack-tox-pep8 https://logs.opendev.org/95/673995/1/gate/openstack-tox-pep8/fbeefac/ : SUCCESS in 4m 46s\n- openstack-tox-py27 https://logs.opendev.org/95/673995/1/gate/openstack-tox-py27/e557a1a/ : SUCCESS in 4m 23s\n- openstack-tox-py35 https://logs.opendev.org/95/673995/1/gate/openstack-tox-py35/89430b3/ : SUCCESS in 4m 56s\n- openstack-tox-py36 https://logs.opendev.org/95/673995/1/gate/openstack-tox-py36/5437b2c/ : SUCCESS in 4m 55s\n- openstack-tox-docs https://logs.opendev.org/95/673995/1/gate/openstack-tox-docs/78346d1/html/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://logs.opendev.org/95/673995/1/gate/openstack-tox-lower-constraints/c1d4e5b/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"02539e596af01e9f8ea0f03441196504a3fda840","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:42:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87c293917e081f106b202eeb986a0316139752e0","revisions":{"87c293917e081f106b202eeb986a0316139752e0":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:01:17.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/95/673995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/95/673995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/673995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/673995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/673995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/95/673995/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-01 19:19:28.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:01:17.000000000","tz":0},"subject":"New validation: ensure services status","message":"New validation: ensure services status\n\nThis new validation ensures we don\u0027t have any failed systemd unit.\n\nFailed units might lead to unexpected errors when we want to update\nor upgrade the system, so we have to ensure we don\u0027t have anything\nin a failed state.\n\nSince this check is pretty generic and useful, it can also be called\nas a post-deploy, post-upgrade task. This ensures we have a corret\nstate AFTER such actions as well.\n\nChange-Id: Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf\n(cherry picked from commit 1265e2f6675b95bf7487fe041066d8de3164e39e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/87c293917e081f106b202eeb986a0316139752e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/87c293917e081f106b202eeb986a0316139752e0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
