)]}'
{"id":"openstack%2Ftripleo-validations~673996","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I8abe0b2fc8b3619c227da01d145355c2d8d1d15d","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule/undercloud-debug-stable/stein","hashtags":[],"change_id":"I8abe0b2fc8b3619c227da01d145355c2d8d1d15d","subject":"Add molecule testing on undercloud-debug role","status":"MERGED","created":"2019-08-01 11:03:33.000000000","updated":"2019-08-06 15:54:05.000000000","submitted":"2019-08-06 15:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673996-1565106845188-72314eed","meta_rev_id":"dc4e6d169ccb1177db43456e8b51a83b2a767a6f","_number":673996,"virtual_id_number":673996,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-06 15:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-06 05:42:56.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-04 12:16:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-04 12:16:33.000000000","_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"date":"2019-08-04 12:16:27.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-06 13:03:29.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-04 12:16:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-04 12:16:27.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-04 12:16:33.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-06 05:42:56.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-06 13:03:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-06 15:54:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7171f5bbb83aafcd11d5e0196702f0f9f067474b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:03:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d22d2dbc94f92919931a5de1e1a7bc0ddb46b022","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:52:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e559a42e7758030274312728923649302cc58dfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:07:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/96/673996/2/check/tripleo-ci-centos-7-containers-multinode/91058de/ : SUCCESS in 2h 14m 38s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/96/673996/2/check/tripleo-ci-centos-7-undercloud-containers/32d9c0e/ : SUCCESS in 1h 35m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/96/673996/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c771618/ : SUCCESS in 2h 23m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/96/673996/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0fc5e46/ : SUCCESS in 2h 45m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/96/673996/2/check/tripleo-ci-centos-7-standalone/2ddae37/ : SUCCESS in 1h 55m 17s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/96/673996/2/check/tripleo-ci-fedora-28-standalone/84809ff/ : SUCCESS in 1h 34m 17s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/96/673996/2/check/openstack-tox-pep8/1e7bfbf/ : SUCCESS in 7m 29s\n- openstack-tox-py27 https://logs.opendev.org/96/673996/2/check/openstack-tox-py27/051c200/ : SUCCESS in 4m 37s\n- openstack-tox-py35 https://logs.opendev.org/96/673996/2/check/openstack-tox-py35/3e0b380/ : SUCCESS in 4m 43s\n- openstack-tox-py36 https://logs.opendev.org/96/673996/2/check/openstack-tox-py36/bbbd88b/ : SUCCESS in 5m 59s\n- openstack-tox-molecule https://logs.opendev.org/96/673996/2/check/openstack-tox-molecule/e8f12b7/tox/reports.html : SUCCESS in 12m 19s\n- openstack-tox-docs https://logs.opendev.org/96/673996/2/check/openstack-tox-docs/d26f49f/html/ : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://logs.opendev.org/96/673996/2/check/openstack-tox-lower-constraints/1252b06/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"73896f0cc7b18adfe59eba981e3998a85edf3f4b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-06 05:42:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0208f87d75e7c16408587852aa52c4fa992c3232","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 13:03:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"14d830f6998feec5e1750506ae0fdd5b3b92bf18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:03:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d3b81cd4f1e404a06834409e3399ff26f95bee4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:54:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/96/673996/2/gate/tripleo-ci-centos-7-containers-multinode/c468f21/ : SUCCESS in 2h 11m 12s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/96/673996/2/gate/tripleo-ci-centos-7-standalone/9ca0dab/ : SUCCESS in 2h 02m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/96/673996/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/11e6afd/ : SUCCESS in 2h 40m 24s\n- openstack-tox-pep8 https://logs.opendev.org/96/673996/2/gate/openstack-tox-pep8/a68cd32/ : SUCCESS in 6m 30s\n- openstack-tox-py27 https://logs.opendev.org/96/673996/2/gate/openstack-tox-py27/4992cf8/ : SUCCESS in 4m 51s\n- openstack-tox-py35 https://logs.opendev.org/96/673996/2/gate/openstack-tox-py35/ce159c6/ : SUCCESS in 5m 34s\n- openstack-tox-py36 https://logs.opendev.org/96/673996/2/gate/openstack-tox-py36/7c749d3/ : SUCCESS in 5m 33s\n- openstack-tox-molecule https://logs.opendev.org/96/673996/2/gate/openstack-tox-molecule/3b40611/tox/reports.html : SUCCESS in 27m 14s\n- openstack-tox-docs https://logs.opendev.org/96/673996/2/gate/openstack-tox-docs/16e1207/html/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/96/673996/2/gate/openstack-tox-lower-constraints/2145934/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b8ffb7a89d4a746941c595892525257f4cd1012c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:54:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e513ca2a04823d65d2529eb8802b52bf44b4105","revisions":{"8b8888988904bc281fd7c11b9d47a9db43a9dbb7":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:03:33.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/96/673996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/96/673996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 11:54:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:03:33.000000000","tz":0},"subject":"Add molecule testing on undercloud-debug role","message":"Add molecule testing on undercloud-debug role\n\nChange-Id: I8abe0b2fc8b3619c227da01d145355c2d8d1d15d\n(cherry picked from commit 95e9eabaa6d4218550278f4b05116bf2de7629d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b8888988904bc281fd7c11b9d47a9db43a9dbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b8888988904bc281fd7c11b9d47a9db43a9dbb7"}]},"branch":"refs/heads/stable/stein"},"3e513ca2a04823d65d2529eb8802b52bf44b4105":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-01 12:52:18.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/96/673996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/96/673996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/96/673996/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 11:54:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:52:16.000000000","tz":0},"subject":"Add molecule testing on undercloud-debug role","message":"Add molecule testing on undercloud-debug role\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: I8abe0b2fc8b3619c227da01d145355c2d8d1d15d\n(cherry picked from commit 95e9eabaa6d4218550278f4b05116bf2de7629d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3e513ca2a04823d65d2529eb8802b52bf44b4105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3e513ca2a04823d65d2529eb8802b52bf44b4105"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
