)]}'
{"id":"openstack%2Ftripleo-validations~673998","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I2e94b72080b73dc37eefd072217392f34e1291f0","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule/controller-ulimits-stable/stein","hashtags":[],"change_id":"I2e94b72080b73dc37eefd072217392f34e1291f0","subject":"Adds molecule testing for controller-ulimits","status":"MERGED","created":"2019-08-01 11:04:33.000000000","updated":"2019-08-02 02:25:27.000000000","submitted":"2019-08-02 02:25:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673998-1564712727588-f6c17697","meta_rev_id":"517df8827277125d7ce5b63def28baa385a986bb","_number":673998,"virtual_id_number":673998,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 02:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 23:08:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-02 02:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-01 23:08:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:04:33.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:04:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 23:08:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 02:10:28.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-08-02 02:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bfa367e69bed9e65d226e29180a98535eb1653c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:04:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5aad8149b2b1bf1623d771ffe006578aa59fde4e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:52:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7000703940a9e8f3939fb3b613fed5e3173e5f15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:00:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/98/673998/2/check/tripleo-ci-centos-7-containers-multinode/478f6b3/ : SUCCESS in 2h 14m 20s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/98/673998/2/check/tripleo-ci-centos-7-undercloud-containers/3e5994e/ : SUCCESS in 1h 39m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/98/673998/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/852f36a/ : SUCCESS in 2h 30m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/98/673998/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/86d9834/ : SUCCESS in 2h 49m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/98/673998/2/check/tripleo-ci-centos-7-standalone/59ca66c/ : SUCCESS in 2h 11m 43s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/98/673998/2/check/tripleo-ci-fedora-28-standalone/64df3bb/ : SUCCESS in 1h 41m 57s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/98/673998/2/check/openstack-tox-pep8/dab3dbc/ : SUCCESS in 5m 44s\n- openstack-tox-py27 https://logs.opendev.org/98/673998/2/check/openstack-tox-py27/df1b03d/ : SUCCESS in 5m 28s\n- openstack-tox-py35 https://logs.opendev.org/98/673998/2/check/openstack-tox-py35/620848a/ : SUCCESS in 5m 07s\n- openstack-tox-py36 https://logs.opendev.org/98/673998/2/check/openstack-tox-py36/ed53233/ : SUCCESS in 4m 51s\n- openstack-tox-molecule https://logs.opendev.org/98/673998/2/check/openstack-tox-molecule/a45c996/tox/reports.html : SUCCESS in 9m 00s\n- openstack-tox-docs https://logs.opendev.org/98/673998/2/check/openstack-tox-docs/a96d262/html/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673998/2/check/openstack-tox-lower-constraints/6a87d8b/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9febbbeec2461e47221c5fa7da249bc660ee680d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 23:08:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"80bba2077db00ad874373c13fc7b89359f6d7560","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:08:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4bf817afe3d1fe3a3e9c0062b41c1063009580e0","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-08-02 02:10:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c7ce4f8712328adabcb0c69f489dff38a8c6f33f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:25:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/98/673998/2/gate/tripleo-ci-centos-7-containers-multinode/328a2a4/ : SUCCESS in 2h 06m 14s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/98/673998/2/gate/tripleo-ci-centos-7-standalone/b779f82/ : SUCCESS in 1h 46m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/98/673998/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7919627/ : SUCCESS in 3h 07m 11s\n- openstack-tox-pep8 https://logs.opendev.org/98/673998/2/gate/openstack-tox-pep8/7a09ee4/ : SUCCESS in 5m 05s\n- openstack-tox-py27 https://logs.opendev.org/98/673998/2/gate/openstack-tox-py27/c2e8431/ : SUCCESS in 5m 09s\n- openstack-tox-py35 https://logs.opendev.org/98/673998/2/gate/openstack-tox-py35/6fb7c8f/ : SUCCESS in 5m 12s\n- openstack-tox-py36 https://logs.opendev.org/98/673998/2/gate/openstack-tox-py36/a09144b/ : SUCCESS in 4m 22s\n- openstack-tox-molecule https://logs.opendev.org/98/673998/2/gate/openstack-tox-molecule/bfb6ed1/tox/reports.html : SUCCESS in 18m 27s\n- openstack-tox-docs https://logs.opendev.org/98/673998/2/gate/openstack-tox-docs/f7302ef/html/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673998/2/gate/openstack-tox-lower-constraints/236c5d1/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d86a36260da0335e793f37c4c24ff729a2a856b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:25:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7db8b3076b7e495ac6492cf99552e970fc6fb5e","revisions":{"0f04f3710b6730953ed9441a3a8dbf99bffad68a":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:04:33.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/98/673998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/673998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 09:25:45.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:04:33.000000000","tz":0},"subject":"Adds molecule testing for controller-ulimits","message":"Adds molecule testing for controller-ulimits\n\nChange-Id: I2e94b72080b73dc37eefd072217392f34e1291f0\n(cherry picked from commit e1d0bde7dd120f42fe1d6301cbdc65e3327615af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0f04f3710b6730953ed9441a3a8dbf99bffad68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0f04f3710b6730953ed9441a3a8dbf99bffad68a"}]},"branch":"refs/heads/stable/stein"},"b7db8b3076b7e495ac6492cf99552e970fc6fb5e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-01 12:52:52.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/98/673998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/673998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/673998/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 09:25:45.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:52:47.000000000","tz":0},"subject":"Adds molecule testing for controller-ulimits","message":"Adds molecule testing for controller-ulimits\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: I2e94b72080b73dc37eefd072217392f34e1291f0\n(cherry picked from commit e1d0bde7dd120f42fe1d6301cbdc65e3327615af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b7db8b3076b7e495ac6492cf99552e970fc6fb5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b7db8b3076b7e495ac6492cf99552e970fc6fb5e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
