)]}'
{"id":"openstack%2Ftripleo-validations~673999","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ie3226b708665b1733891d592a54cc6f2398899df","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule/controller-token-stable/stein","hashtags":[],"change_id":"Ie3226b708665b1733891d592a54cc6f2398899df","subject":"Adds molecule testing for controller-token","status":"MERGED","created":"2019-08-01 11:05:06.000000000","updated":"2019-08-02 01:45:39.000000000","submitted":"2019-08-02 01:45:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673999-1564710339777-b398d57e","meta_rev_id":"1b56d91c912b205db511128256f415b568905c26","_number":673999,"virtual_id_number":673999,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 01:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 23:07:59.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-01 23:07:59.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:05:06.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:05:19.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 23:07:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 01:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a49f4ff999d68f52823050ac1a55a97146ca9d30","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:05:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"94f81d45778cf1a551e93a2ac2c01ae0623b12d9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:53:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0bdbc8a57f740e189292ad8fedb619ed4eaaa8e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:55:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/99/673999/2/check/tripleo-ci-centos-7-containers-multinode/d409f0e/ : SUCCESS in 2h 08m 24s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/99/673999/2/check/tripleo-ci-centos-7-undercloud-containers/c5fba20/ : SUCCESS in 1h 39m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/99/673999/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6781e9a/ : SUCCESS in 2h 16m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/99/673999/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ca94999/ : SUCCESS in 2h 43m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/99/673999/2/check/tripleo-ci-centos-7-standalone/0fa4122/ : SUCCESS in 1h 51m 33s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/99/673999/2/check/tripleo-ci-fedora-28-standalone/0c9ab87/ : SUCCESS in 1h 45m 53s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/99/673999/2/check/openstack-tox-pep8/75244cd/ : SUCCESS in 4m 12s\n- openstack-tox-py27 https://logs.opendev.org/99/673999/2/check/openstack-tox-py27/61212ff/ : SUCCESS in 4m 12s\n- openstack-tox-py35 https://logs.opendev.org/99/673999/2/check/openstack-tox-py35/716dff8/ : SUCCESS in 7m 56s\n- openstack-tox-py36 https://logs.opendev.org/99/673999/2/check/openstack-tox-py36/d5750e2/ : SUCCESS in 5m 33s\n- openstack-tox-molecule https://logs.opendev.org/99/673999/2/check/openstack-tox-molecule/d0a2f86/tox/reports.html : SUCCESS in 12m 36s\n- openstack-tox-docs https://logs.opendev.org/99/673999/2/check/openstack-tox-docs/00bf163/html/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://logs.opendev.org/99/673999/2/check/openstack-tox-lower-constraints/b25a488/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"706f65dbaf1c4c59febec2774cef66003c82c4cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 23:07:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"741bc7262457cb98babaee04c198d48028fca0e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:08:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16c1f9e8777d6db1e072d2d3321848defc28c9a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:45:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/99/673999/2/gate/tripleo-ci-centos-7-containers-multinode/2832954/ : SUCCESS in 2h 16m 08s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/99/673999/2/gate/tripleo-ci-centos-7-standalone/5252cc2/ : SUCCESS in 1h 51m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/99/673999/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b67e016/ : SUCCESS in 2h 17m 11s\n- openstack-tox-pep8 https://logs.opendev.org/99/673999/2/gate/openstack-tox-pep8/0062f90/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/99/673999/2/gate/openstack-tox-py27/371f28a/ : SUCCESS in 4m 14s\n- openstack-tox-py35 https://logs.opendev.org/99/673999/2/gate/openstack-tox-py35/f19bf74/ : SUCCESS in 4m 36s\n- openstack-tox-py36 https://logs.opendev.org/99/673999/2/gate/openstack-tox-py36/acd8362/ : SUCCESS in 4m 17s\n- openstack-tox-molecule https://logs.opendev.org/99/673999/2/gate/openstack-tox-molecule/1889121/tox/reports.html : SUCCESS in 17m 17s\n- openstack-tox-docs https://logs.opendev.org/99/673999/2/gate/openstack-tox-docs/0493c50/html/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/99/673999/2/gate/openstack-tox-lower-constraints/73b891c/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"55583e321a004bbe1a558b7458fb9febca47421f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:45:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0386bc224ab3a921cee56ef1372a038aa90a528e","revisions":{"86e200d5fc1fc19356e48eec7801ec5d574c2df3":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:05:06.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/99/673999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/99/673999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 08:55:54.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:05:06.000000000","tz":0},"subject":"Adds molecule testing for controller-token","message":"Adds molecule testing for controller-token\n\nChange-Id: Ie3226b708665b1733891d592a54cc6f2398899df\n(cherry picked from commit 1ace29c69400173e05346dbc4906e003f72899b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/86e200d5fc1fc19356e48eec7801ec5d574c2df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/86e200d5fc1fc19356e48eec7801ec5d574c2df3"}]},"branch":"refs/heads/stable/stein"},"0386bc224ab3a921cee56ef1372a038aa90a528e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-01 12:53:25.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/99/673999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/99/673999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/99/673999/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 08:55:54.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:53:21.000000000","tz":0},"subject":"Adds molecule testing for controller-token","message":"Adds molecule testing for controller-token\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: Ie3226b708665b1733891d592a54cc6f2398899df\n(cherry picked from commit 1ace29c69400173e05346dbc4906e003f72899b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0386bc224ab3a921cee56ef1372a038aa90a528e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0386bc224ab3a921cee56ef1372a038aa90a528e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
