)]}'
{"id":"openstack%2Ftripleo-validations~674000","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule/dns-stable/stein","hashtags":[],"change_id":"Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da","subject":"Adds molecule testing for DNS validation","status":"MERGED","created":"2019-08-01 11:05:34.000000000","updated":"2019-08-02 02:00:27.000000000","submitted":"2019-08-02 02:00:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674000-1564711227309-c33df9d0","meta_rev_id":"5501d67d403176df6d4abcb741640328a67d34cf","_number":674000,"virtual_id_number":674000,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 02:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 23:08:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-01 23:08:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:05:34.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:05:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 23:08:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 02:00:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4cb0a04fe23e9f83347139343ed5a4886d3fd50","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:05:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fdfc48b186d7742484eaf0a29daf1de00f31d564","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:53:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8de1c4e5cbc7865180fb1c65374a0dd129881b7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:56:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/00/674000/2/check/tripleo-ci-centos-7-containers-multinode/8615245/ : SUCCESS in 2h 14m 45s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/00/674000/2/check/tripleo-ci-centos-7-undercloud-containers/afcbf3d/ : SUCCESS in 1h 47m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/00/674000/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8e3bfd8/ : SUCCESS in 2h 32m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/00/674000/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/92a487e/ : SUCCESS in 2h 30m 55s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/00/674000/2/check/tripleo-ci-centos-7-standalone/1aa8834/ : SUCCESS in 1h 53m 49s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/00/674000/2/check/tripleo-ci-fedora-28-standalone/28a3d31/ : SUCCESS in 2h 15m 28s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/00/674000/2/check/openstack-tox-pep8/d7c1b1e/ : SUCCESS in 4m 07s\n- openstack-tox-py27 https://logs.opendev.org/00/674000/2/check/openstack-tox-py27/8dc08d1/ : SUCCESS in 3m 49s\n- openstack-tox-py35 https://logs.opendev.org/00/674000/2/check/openstack-tox-py35/250ba53/ : SUCCESS in 4m 14s\n- openstack-tox-py36 https://logs.opendev.org/00/674000/2/check/openstack-tox-py36/868b655/ : SUCCESS in 3m 46s\n- openstack-tox-molecule https://logs.opendev.org/00/674000/2/check/openstack-tox-molecule/00de613/tox/reports.html : SUCCESS in 9m 49s\n- openstack-tox-docs https://logs.opendev.org/00/674000/2/check/openstack-tox-docs/2e0e9d1/html/ : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://logs.opendev.org/00/674000/2/check/openstack-tox-lower-constraints/39a5429/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f00e5e7a68696cb804b4cc11309871ce29fdebb4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 23:08:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dd18904e197c09d2ca93a760d29829244ef78e13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:08:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82b5fc7e9c95247b7822c06f8b75973f87881175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:00:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/00/674000/2/gate/tripleo-ci-centos-7-containers-multinode/0d85095/ : SUCCESS in 2h 26m 21s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/00/674000/2/gate/tripleo-ci-centos-7-standalone/17c25dd/ : SUCCESS in 1h 58m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/00/674000/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/08bc72b/ : SUCCESS in 2h 32m 23s\n- openstack-tox-pep8 https://logs.opendev.org/00/674000/2/gate/openstack-tox-pep8/730454b/ : SUCCESS in 5m 47s\n- openstack-tox-py27 https://logs.opendev.org/00/674000/2/gate/openstack-tox-py27/04d0678/ : SUCCESS in 4m 28s\n- openstack-tox-py35 https://logs.opendev.org/00/674000/2/gate/openstack-tox-py35/cbb0b37/ : SUCCESS in 5m 45s\n- openstack-tox-py36 https://logs.opendev.org/00/674000/2/gate/openstack-tox-py36/347dfc0/ : SUCCESS in 4m 49s\n- openstack-tox-molecule https://logs.opendev.org/00/674000/2/gate/openstack-tox-molecule/679e5ff/tox/reports.html : SUCCESS in 15m 58s\n- openstack-tox-docs https://logs.opendev.org/00/674000/2/gate/openstack-tox-docs/0a5c9fd/html/ : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/00/674000/2/gate/openstack-tox-lower-constraints/32e364e/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9edd5da71c3237b1325687a273858646f54cc096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:00:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ac7e8d392a76f1712c554822ba3e22d52f173df","revisions":{"9b57fad89464e65cb26dcda7d4ea8105511615df":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:05:34.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/674000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/674000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 09:48:17.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:05:34.000000000","tz":0},"subject":"Adds molecule testing for DNS validation","message":"Adds molecule testing for DNS validation\n\nChange-Id: Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da\n(cherry picked from commit 647e9afc264fab5a0672d372d7244a171104486e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b57fad89464e65cb26dcda7d4ea8105511615df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b57fad89464e65cb26dcda7d4ea8105511615df"}]},"branch":"refs/heads/stable/stein"},"8ac7e8d392a76f1712c554822ba3e22d52f173df":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-01 12:53:43.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/674000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/674000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/674000/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 09:48:17.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:53:41.000000000","tz":0},"subject":"Adds molecule testing for DNS validation","message":"Adds molecule testing for DNS validation\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: Ia1dba94328e885864233ca4eff0fb9aa7e7ce2da\n(cherry picked from commit 647e9afc264fab5a0672d372d7244a171104486e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8ac7e8d392a76f1712c554822ba3e22d52f173df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8ac7e8d392a76f1712c554822ba3e22d52f173df"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
