)]}'
{"id":"openstack%2Ftripleo-validations~674001","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I20b55dc55b833a87930a4d9efd9fe81d4759ccb5","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule/undercloud-disk-space-stable/stein","hashtags":[],"change_id":"I20b55dc55b833a87930a4d9efd9fe81d4759ccb5","subject":"Adds molecule tests for undercloud-disk-space","status":"MERGED","created":"2019-08-01 11:06:03.000000000","updated":"2019-08-02 01:45:39.000000000","submitted":"2019-08-02 01:45:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674001-1564710339076-e5490da1","meta_rev_id":"68a5643ec801bdbee2840d25ddaa283673595a02","_number":674001,"virtual_id_number":674001,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 01:45:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 23:07:34.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-01 23:07:34.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:06:03.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:06:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 23:07:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 01:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d3ff5b7b3c0e4d9853ef0e8ff60226127b7c34e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:06:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2c12191d2530a381901b00f276a2689f2c4db29f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:53:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"76335fef3c2b6b6e32606bc30ebb7627ed2b9f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/674001/2/check/tripleo-ci-centos-7-containers-multinode/778b5a1/ : SUCCESS in 2h 30m 48s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/01/674001/2/check/tripleo-ci-centos-7-undercloud-containers/5fd5a24/ : SUCCESS in 1h 44m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/01/674001/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8ca6912/ : SUCCESS in 2h 18m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/01/674001/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6cbcd1d/ : SUCCESS in 2h 30m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/674001/2/check/tripleo-ci-centos-7-standalone/c28c1d1/ : SUCCESS in 2h 01m 28s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/01/674001/2/check/tripleo-ci-fedora-28-standalone/673d59a/ : SUCCESS in 1h 48m 44s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/01/674001/2/check/openstack-tox-pep8/c18df6a/ : SUCCESS in 5m 41s\n- openstack-tox-py27 https://logs.opendev.org/01/674001/2/check/openstack-tox-py27/0295c8a/ : SUCCESS in 4m 17s\n- openstack-tox-py35 https://logs.opendev.org/01/674001/2/check/openstack-tox-py35/4c560b9/ : SUCCESS in 5m 11s\n- openstack-tox-py36 https://logs.opendev.org/01/674001/2/check/openstack-tox-py36/e30639c/ : SUCCESS in 5m 01s\n- openstack-tox-molecule https://logs.opendev.org/01/674001/2/check/openstack-tox-molecule/c3b83a3/tox/reports.html : SUCCESS in 10m 03s\n- openstack-tox-docs https://logs.opendev.org/01/674001/2/check/openstack-tox-docs/d116797/html/ : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/01/674001/2/check/openstack-tox-lower-constraints/dc28a84/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0f84ad5a0082630a84c9e85a02398ca1e224c926","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 23:07:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86461773a727d89d73be59d3486462891393427d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:07:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70d926ae3b8e5da0cd4cf0f94bc55b6c35b4e9b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:45:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/674001/2/gate/tripleo-ci-centos-7-containers-multinode/ef8f705/ : SUCCESS in 2h 15m 02s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/674001/2/gate/tripleo-ci-centos-7-standalone/2f67398/ : SUCCESS in 1h 57m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/01/674001/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/714aa2c/ : SUCCESS in 2h 32m 09s\n- openstack-tox-pep8 https://logs.opendev.org/01/674001/2/gate/openstack-tox-pep8/6dd887f/ : SUCCESS in 5m 57s\n- openstack-tox-py27 https://logs.opendev.org/01/674001/2/gate/openstack-tox-py27/d656c8f/ : SUCCESS in 6m 08s\n- openstack-tox-py35 https://logs.opendev.org/01/674001/2/gate/openstack-tox-py35/cc2e90f/ : SUCCESS in 6m 33s\n- openstack-tox-py36 https://logs.opendev.org/01/674001/2/gate/openstack-tox-py36/6d99e65/ : SUCCESS in 5m 23s\n- openstack-tox-molecule https://logs.opendev.org/01/674001/2/gate/openstack-tox-molecule/2bc3bd5/tox/reports.html : SUCCESS in 13m 57s\n- openstack-tox-docs https://logs.opendev.org/01/674001/2/gate/openstack-tox-docs/09735c3/html/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/01/674001/2/gate/openstack-tox-lower-constraints/67fe80e/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8b296713f64255f21e3b9e43401684e2372a96b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:45:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"110d27f4e702aa9e7a4658defc68455674ee013e","revisions":{"a95ad30c28db5551c8a57d738cd328ceda8765ed":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:06:03.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/01/674001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/01/674001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 14:05:22.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:06:03.000000000","tz":0},"subject":"Adds molecule tests for undercloud-disk-space","message":"Adds molecule tests for undercloud-disk-space\n\nChange-Id: I20b55dc55b833a87930a4d9efd9fe81d4759ccb5\n(cherry picked from commit 5adfef9d9a61402617591ef58315b9cd62f60907)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a95ad30c28db5551c8a57d738cd328ceda8765ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a95ad30c28db5551c8a57d738cd328ceda8765ed"}]},"branch":"refs/heads/stable/stein"},"110d27f4e702aa9e7a4658defc68455674ee013e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-01 12:53:59.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/01/674001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/01/674001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/01/674001/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-22 14:05:22.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:53:57.000000000","tz":0},"subject":"Adds molecule tests for undercloud-disk-space","message":"Adds molecule tests for undercloud-disk-space\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: I20b55dc55b833a87930a4d9efd9fe81d4759ccb5\n(cherry picked from commit 5adfef9d9a61402617591ef58315b9cd62f60907)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/110d27f4e702aa9e7a4658defc68455674ee013e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/110d27f4e702aa9e7a4658defc68455674ee013e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
