)]}'
{"id":"openstack%2Ftripleo-validations~674002","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ib8ae41451f245d4e2d40864cc36515a07a1878bf","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"molecule-stable-stein","hashtags":[],"change_id":"Ib8ae41451f245d4e2d40864cc36515a07a1878bf","subject":"Adds molecule tests for ctlplane-ip-range validation","status":"MERGED","created":"2019-08-01 11:07:25.000000000","updated":"2019-08-02 16:33:24.000000000","submitted":"2019-08-02 16:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674002-1564763604818-3342d8ab","meta_rev_id":"7c346dd8601006bda4849e3626cad22be6bfae2a","_number":674002,"virtual_id_number":674002,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-02 16:33:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-02 13:22:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-02 13:22:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:07:25.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-01 11:07:35.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-02 13:22:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 16:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99dd4b2c287eb5931fb5350589bf1378e1d6602d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:07:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dc71f28bcc9eb01caa954b49916f59060c9cec60","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:10:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cce8f380ed054bce9683c210d1653ce528858b97","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 12:54:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1f20bad97c993786c0d6d78f11341ac3d027b105","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 14:02:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30f5b56437375e0905f380ae1b7d0bb2cf74528f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 00:33:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/02/674002/4/check/tripleo-ci-centos-7-containers-multinode/fcd3103/ : SUCCESS in 2h 14m 05s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/02/674002/4/check/tripleo-ci-centos-7-undercloud-containers/8613c0f/ : SUCCESS in 1h 37m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/02/674002/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/208595d/ : SUCCESS in 2h 35m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/02/674002/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2bd2b06/ : SUCCESS in 2h 50m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/02/674002/4/check/tripleo-ci-centos-7-standalone/0cedd3b/ : SUCCESS in 1h 57m 16s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/02/674002/4/check/tripleo-ci-fedora-28-standalone/cfa44eb/ : SUCCESS in 1h 50m 21s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/02/674002/4/check/openstack-tox-pep8/aabfa88/ : SUCCESS in 3m 53s\n- openstack-tox-py27 https://logs.opendev.org/02/674002/4/check/openstack-tox-py27/32b14a1/ : SUCCESS in 4m 01s\n- openstack-tox-py35 https://logs.opendev.org/02/674002/4/check/openstack-tox-py35/31cf849/ : SUCCESS in 3m 49s\n- openstack-tox-py36 https://logs.opendev.org/02/674002/4/check/openstack-tox-py36/2c67244/ : SUCCESS in 3m 49s\n- openstack-tox-molecule https://logs.opendev.org/02/674002/4/check/openstack-tox-molecule/f84a7ac/tox/reports.html : SUCCESS in 13m 51s\n- openstack-tox-docs https://logs.opendev.org/02/674002/4/check/openstack-tox-docs/abcae90/html/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/02/674002/4/check/openstack-tox-lower-constraints/eff411c/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"fc9c80632b1211ef3f1432683723211e33af2049","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 13:22:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb3f69efaf27def56ffb55b42bed1354a589d83a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 13:23:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c51f4397d2f2dc4de67bb2c7fa04571f68e1595c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7aff3b876f6f64cfdab49fdea8c0bbadbd6cda95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:33:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/02/674002/4/gate/tripleo-ci-centos-7-containers-multinode/e655ede/ : SUCCESS in 2h 39m 58s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/02/674002/4/gate/tripleo-ci-centos-7-standalone/4219c89/ : SUCCESS in 2h 15m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/02/674002/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/896566e/ : SUCCESS in 2h 39m 21s\n- openstack-tox-pep8 https://logs.opendev.org/02/674002/4/gate/openstack-tox-pep8/92bb360/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/02/674002/4/gate/openstack-tox-py27/9ef0809/ : SUCCESS in 4m 11s\n- openstack-tox-py35 https://logs.opendev.org/02/674002/4/gate/openstack-tox-py35/a09844a/ : SUCCESS in 5m 10s\n- openstack-tox-py36 https://logs.opendev.org/02/674002/4/gate/openstack-tox-py36/6dcce0d/ : SUCCESS in 4m 59s\n- openstack-tox-molecule https://logs.opendev.org/02/674002/4/gate/openstack-tox-molecule/b1127ce/tox/reports.html : SUCCESS in 22m 42s\n- openstack-tox-docs https://logs.opendev.org/02/674002/4/gate/openstack-tox-docs/3aab8ff/html/ : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/02/674002/4/gate/openstack-tox-lower-constraints/d8e0d97/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d3c81eeff56cc214b2fe780f3520ab64f548cc6","revisions":{"d3cd7c726596dbcfc99c4927e29dd3e44b74084c":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:07:25.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/02/674002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/02/674002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/1"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:07:25.000000000","tz":0},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n(cherry picked from commit 49aa25278db5031d0543169ad2e12676ca89fc7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3cd7c726596dbcfc99c4927e29dd3e44b74084c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3cd7c726596dbcfc99c4927e29dd3e44b74084c"}]},"branch":"refs/heads/stable/stein"},"2819c5a0b730080074bc67ae837a7bb13d300bf9":{"kind":"NO_CHANGE","_number":2,"created":"2019-08-01 11:10:19.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/02/674002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/02/674002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/2"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 11:09:42.000000000","tz":120},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n(cherry picked from commit 49aa25278db5031d0543169ad2e12676ca89fc7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2819c5a0b730080074bc67ae837a7bb13d300bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2819c5a0b730080074bc67ae837a7bb13d300bf9"}]},"branch":"refs/heads/stable/stein"},"083a6a2b97feb9e28af4bf85c03e78efe79f66b1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-01 12:54:26.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/02/674002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/02/674002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/3"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 12:54:22.000000000","tz":0},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n(cherry picked from commit 49aa25278db5031d0543169ad2e12676ca89fc7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/083a6a2b97feb9e28af4bf85c03e78efe79f66b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/083a6a2b97feb9e28af4bf85c03e78efe79f66b1"}]},"branch":"refs/heads/stable/stein"},"6d3c81eeff56cc214b2fe780f3520ab64f548cc6":{"kind":"REWORK","_number":4,"created":"2019-08-01 14:02:47.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/02/674002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/02/674002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/02/674002/4"}}},"commit":{"parents":[{"commit":"23ebc8ae7317a66205157519628c23a5f63ab066","subject":"Support multi-arch in deploy image validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/23ebc8ae7317a66205157519628c23a5f63ab066"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-07-23 06:03:06.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-01 14:01:47.000000000","tz":120},"subject":"Adds molecule tests for ctlplane-ip-range validation","message":"Adds molecule tests for ctlplane-ip-range validation\n\nDepends-On: I0551ab56324ede7f1d33e0e9a680c26ea164023c\nChange-Id: Ib8ae41451f245d4e2d40864cc36515a07a1878bf\n(cherry picked from commit 49aa25278db5031d0543169ad2e12676ca89fc7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d3c81eeff56cc214b2fe780f3520ab64f548cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d3c81eeff56cc214b2fe780f3520ab64f548cc6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
