)]}'
{"id":"openstack%2Ftripleo-validations~674217","triplet_id":"openstack%2Ftripleo-validations~master~Ic17307efc2b4907cc21df8289a2e0c84d2f30b48","project":"openstack/tripleo-validations","branch":"master","hashtags":[],"change_id":"Ic17307efc2b4907cc21df8289a2e0c84d2f30b48","subject":"Improve Ceph health checks to work with custom cluster names","status":"MERGED","created":"2019-08-02 08:09:09.000000000","updated":"2019-08-06 10:36:25.000000000","submitted":"2019-08-05 20:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674217-1565036347111-efd57ca1","meta_rev_id":"e298ec4382025377c768860ee65019a04fc0f91b","_number":674217,"virtual_id_number":674217,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-05 20:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":2,"date":"2019-08-05 14:48:38.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":2,"date":"2019-08-02 09:18:33.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-08-02 09:20:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":1,"date":"2019-08-05 14:48:38.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 09:18:33.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-02 09:20:03.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-05 14:48:38.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-08-05 20:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ea195b35cdeba6fb0cc9091b0782556af6f21bf","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-02 08:09:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a88e1837383857e4ca19c6ea6383d87f906504f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-02 09:18:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c390140a05a319a092b792b4eb0d0f90a4b02860","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-02 09:20:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd9ef510fa5e4bc00d12cc3bcc19148a507e9251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:03:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-containers-multinode/e29d897/ : SUCCESS in 2h 47m 56s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-undercloud-containers/819f038/ : SUCCESS in 1h 31m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/37f270a/ : SUCCESS in 2h 22m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/723672e/ : SUCCESS in 2h 31m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-standalone/7957c3e/ : SUCCESS in 1h 50m 09s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/17/674217/1/check/tripleo-ci-fedora-28-standalone/7aadaf6/ : FAILURE in 59m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/17/674217/1/check/tripleo-ci-centos-7-standalone-upgrade/5eecedd/ : FAILURE in 1h 36m 23s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/17/674217/1/check/openstack-tox-pep8/d3e36f4/ : SUCCESS in 6m 52s\n- openstack-tox-py27 https://logs.opendev.org/17/674217/1/check/openstack-tox-py27/ec3f784/ : SUCCESS in 5m 43s\n- openstack-tox-py36 https://logs.opendev.org/17/674217/1/check/openstack-tox-py36/714a16d/ : SUCCESS in 4m 19s\n- openstack-tox-py37 https://logs.opendev.org/17/674217/1/check/openstack-tox-py37/c4d331a/ : SUCCESS in 5m 09s\n- openstack-tox-molecule https://logs.opendev.org/17/674217/1/check/openstack-tox-molecule/543fa4f/tox/reports.html : SUCCESS in 29m 30s\n- openstack-tox-docs https://logs.opendev.org/17/674217/1/check/openstack-tox-docs/96f0b33/html/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://logs.opendev.org/17/674217/1/check/openstack-tox-lower-constraints/d09c7fe/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bad305e7596478e32a4f6b13db8970a0e74814cf","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-08-05 14:48:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nnice. thanks","accounts_in_message":[],"_revision_number":1},{"id":"fb76dfd643ffaa192b84257ef94b5c6742889c99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:48:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13ff9984d3cea533a9c37934a6908980cf530b3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/17/674217/1/gate/tripleo-ci-centos-7-containers-multinode/b36c3f1/ : SUCCESS in 2h 33m 40s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/17/674217/1/gate/tripleo-ci-centos-7-standalone/61b0740/ : SUCCESS in 1h 55m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/17/674217/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/94dadd3/ : SUCCESS in 2h 16m 24s\n- openstack-tox-pep8 https://logs.opendev.org/17/674217/1/gate/openstack-tox-pep8/f43791b/ : SUCCESS in 5m 08s\n- openstack-tox-py27 https://logs.opendev.org/17/674217/1/gate/openstack-tox-py27/dd52ca9/ : SUCCESS in 3m 43s\n- openstack-tox-py36 https://logs.opendev.org/17/674217/1/gate/openstack-tox-py36/2ab3c25/ : SUCCESS in 4m 10s\n- openstack-tox-py37 https://logs.opendev.org/17/674217/1/gate/openstack-tox-py37/219c7ae/ : SUCCESS in 4m 13s\n- openstack-tox-molecule https://logs.opendev.org/17/674217/1/gate/openstack-tox-molecule/1a7957f/tox/reports.html : SUCCESS in 25m 30s\n- openstack-tox-docs https://logs.opendev.org/17/674217/1/gate/openstack-tox-docs/2ca1f95/html/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/17/674217/1/gate/openstack-tox-lower-constraints/b662da0/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5bfdac19e822726030adc1054cb5ac8d3889689a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:19:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e298ec4382025377c768860ee65019a04fc0f91b","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-06 10:36:25.000000000","message":"Created a revert of this change as I14bdbbea58dce26993c98f959320dbf063406268","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344","revisions":{"ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344":{"kind":"REWORK","_number":1,"created":"2019-08-02 08:09:09.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/17/674217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/17/674217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/17/674217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/17/674217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/17/674217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/17/674217/1"}}},"commit":{"parents":[{"commit":"0204d7665ccec58d17feae46d9f288a638fad804","subject":"Merge \"Enable fact gathering for needed validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0204d7665ccec58d17feae46d9f288a638fad804"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-02 08:08:06.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-02 08:08:06.000000000","tz":120},"subject":"Improve Ceph health checks to work with custom cluster names","message":"Improve Ceph health checks to work with custom cluster names\n\nWhen the Ceph cluster name is customized the ceph commands need\nto be told the custom cluster name.\n\nChange-Id: Ic17307efc2b4907cc21df8289a2e0c84d2f30b48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
