)]}'
{"id":"openstack%2Ftripleo-validations~674567","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I483d5e7b11376dc0141c506bf07ccd5876e4076e","project":"openstack/tripleo-validations","branch":"stable/stein","hashtags":[],"change_id":"I483d5e7b11376dc0141c506bf07ccd5876e4076e","subject":"Adding new role automation with Ansible","status":"MERGED","created":"2019-08-05 10:26:09.000000000","updated":"2019-08-07 21:06:18.000000000","submitted":"2019-08-07 21:06:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674567-1565211978689-01c2484d","meta_rev_id":"abf7657a7ede0ddcf2b835de360cddffc303029c","_number":674567,"virtual_id_number":674567,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-07 08:11:39.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-08-07 21:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2019-08-07 14:04:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-05 10:26:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-08-05 10:26:37.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":1,"date":"2019-08-07 14:04:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-05 10:26:34.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-05 10:26:27.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 10:26:27.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-05 10:26:34.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-08-05 10:26:37.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-08-05 10:26:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-07 14:04:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-07 21:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f4e6926c5b483c730579dd20bcbe2df76b75fee","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-05 10:26:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03c9db7fc0a19a6d55d29fbb2fa915284468eb19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 12:02:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-containers-multinode/ac9083a/ : FAILURE in 1h 28m 21s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-undercloud-containers/c6611e7/ : FAILURE in 1h 26m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9f86704/ : FAILURE in 1h 18m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b45ed36/ : FAILURE in 1h 28m 12s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-standalone/32a454f/ : FAILURE in 1h 02m 03s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/67/674567/1/check/tripleo-ci-fedora-28-standalone/41e684a/ : FAILURE in 1h 12m 42s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/67/674567/1/check/openstack-tox-pep8/7fe4a33/ : SUCCESS in 4m 54s\n- openstack-tox-py27 https://logs.opendev.org/67/674567/1/check/openstack-tox-py27/ef4a9d9/ : SUCCESS in 5m 01s\n- openstack-tox-py35 https://logs.opendev.org/67/674567/1/check/openstack-tox-py35/35e8cf1/ : SUCCESS in 4m 48s\n- openstack-tox-py36 https://logs.opendev.org/67/674567/1/check/openstack-tox-py36/0f2d2b6/ : SUCCESS in 5m 52s\n- openstack-tox-molecule https://logs.opendev.org/67/674567/1/check/openstack-tox-molecule/076d628/tox/reports.html : SUCCESS in 20m 11s\n- openstack-tox-docs https://logs.opendev.org/67/674567/1/check/openstack-tox-docs/c1aebe3/html/ : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/674567/1/check/openstack-tox-lower-constraints/e24f604/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cf30335945f8ea565762bb473c3ae6faeafc5199","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-07 08:11:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bb3308e3c8704546e641a0426c3661459f60809","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-containers-multinode/87a85e4/ : SUCCESS in 2h 06m 56s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-undercloud-containers/a2daf32/ : SUCCESS in 1h 40m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/85ead1a/ : SUCCESS in 2h 28m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7985f31/ : SUCCESS in 2h 57m 15s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/67/674567/1/check/tripleo-ci-centos-7-standalone/b7b843f/ : SUCCESS in 2h 03m 39s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/67/674567/1/check/tripleo-ci-fedora-28-standalone/9581562/ : SUCCESS in 1h 55m 55s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/67/674567/1/check/openstack-tox-pep8/eed635c/ : SUCCESS in 4m 18s\n- openstack-tox-py27 https://logs.opendev.org/67/674567/1/check/openstack-tox-py27/c97cb5b/ : SUCCESS in 5m 09s\n- openstack-tox-py35 https://logs.opendev.org/67/674567/1/check/openstack-tox-py35/cd4acb7/ : SUCCESS in 4m 53s\n- openstack-tox-py36 https://logs.opendev.org/67/674567/1/check/openstack-tox-py36/0e544e5/ : SUCCESS in 5m 09s\n- openstack-tox-molecule https://logs.opendev.org/67/674567/1/check/openstack-tox-molecule/4989475/tox/reports.html : SUCCESS in 22m 58s\n- openstack-tox-docs https://logs.opendev.org/67/674567/1/check/openstack-tox-docs/583e606/html/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/674567/1/check/openstack-tox-lower-constraints/c68a2ce/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"921209551a280c1507e331ae0cf1cc47515c8509","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 14:04:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5e077fb55ddee9a0e9b6a25fc8883833362bc42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:04:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b17959cb9785d83f7028ca4971a4c844d3f30a6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/67/674567/1/gate/tripleo-ci-centos-7-containers-multinode/9020315/ : SUCCESS in 2h 21m 06s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/67/674567/1/gate/tripleo-ci-centos-7-standalone/bf9e4fe/ : SUCCESS in 2h 05m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/67/674567/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a58b293/ : SUCCESS in 2h 31m 58s\n- openstack-tox-pep8 https://logs.opendev.org/67/674567/1/gate/openstack-tox-pep8/e427fe8/ : SUCCESS in 6m 38s\n- openstack-tox-py27 https://logs.opendev.org/67/674567/1/gate/openstack-tox-py27/5e4c1e7/ : SUCCESS in 4m 36s\n- openstack-tox-py35 https://logs.opendev.org/67/674567/1/gate/openstack-tox-py35/472867d/ : SUCCESS in 5m 47s\n- openstack-tox-py36 https://logs.opendev.org/67/674567/1/gate/openstack-tox-py36/8c53aa4/ : SUCCESS in 4m 54s\n- openstack-tox-molecule https://logs.opendev.org/67/674567/1/gate/openstack-tox-molecule/722369d/tox/reports.html : SUCCESS in 26m 26s\n- openstack-tox-docs https://logs.opendev.org/67/674567/1/gate/openstack-tox-docs/9baea51/html/ : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/674567/1/gate/openstack-tox-lower-constraints/765ad5d/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"abf7657a7ede0ddcf2b835de360cddffc303029c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88641966d31d1d18bd11208214e0f58bae6f2cba","revisions":{"88641966d31d1d18bd11208214e0f58bae6f2cba":{"kind":"REWORK","_number":1,"created":"2019-08-05 10:26:09.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/67/674567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/67/674567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/67/674567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/67/674567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/67/674567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/67/674567/1"}}},"commit":{"parents":[{"commit":"1f0c57fb0979959dc69733b672028ffcc3b695bd","subject":"Merge \"Adds molecule tests for ctlplane-ip-range validation\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1f0c57fb0979959dc69733b672028ffcc3b695bd"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 14:04:23.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 10:15:56.000000000","tz":120},"subject":"Adding new role automation with Ansible","message":"Adding new role automation with Ansible\n\nThis patch adds a basic skeleton to create a new roles including\nmolecule structure.\n\nChange-Id: I483d5e7b11376dc0141c506bf07ccd5876e4076e\nPatch-Inspired-By: tripleo-ansible-Team++\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 9bbd3b26df331ad7307d8f0fe69894277eafdecc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88641966d31d1d18bd11208214e0f58bae6f2cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88641966d31d1d18bd11208214e0f58bae6f2cba"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
