)]}'
{"id":"openstack%2Ftripleo-validations~674641","triplet_id":"openstack%2Ftripleo-validations~master~I4620a7e026bf1ba276d9219630972c14fa0a0137","project":"openstack/tripleo-validations","branch":"master","topic":"containerized-undercloud-podman-refactoring","hashtags":[],"change_id":"I4620a7e026bf1ba276d9219630972c14fa0a0137","subject":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman","status":"ABANDONED","created":"2019-08-05 15:58:45.000000000","updated":"2020-09-18 16:51:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5faaf51e5e84b8646b62fa880560d40954bc3e7","_number":674641,"virtual_id_number":674641,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-08-13 12:54:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 12:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 16:51:18.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"}],"messages":[{"id":"7a15f486314fb3c6e7ffb658603cbab156cad3ff","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-05 15:58:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae9a6dd2fe05048167a5b90eb192e4318ee21486","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-05 16:00:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54ba0139ae1a23c9e9aaff1a136c22c0b199b385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:53:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-containers-multinode/2df3683/ : FAILURE in 30m 42s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-undercloud-containers/6695b4d/ : FAILURE in 23m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b72f754/ : FAILURE in 30m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8c2bb61/ : FAILURE in 28m 28s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-standalone/c9802fe/ : FAILURE in 23m 47s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/41/674641/2/check/tripleo-ci-fedora-28-standalone/58433b0/ : FAILURE in 26m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/41/674641/2/check/tripleo-ci-centos-7-standalone-upgrade/cf56c47/ : FAILURE in 45m 07s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/41/674641/2/check/openstack-tox-pep8/f83152e/ : SUCCESS in 5m 08s\n- openstack-tox-py27 https://logs.opendev.org/41/674641/2/check/openstack-tox-py27/7ae7271/ : SUCCESS in 4m 24s\n- openstack-tox-py36 https://logs.opendev.org/41/674641/2/check/openstack-tox-py36/4d946f4/ : SUCCESS in 4m 35s\n- openstack-tox-py37 https://logs.opendev.org/41/674641/2/check/openstack-tox-py37/2a031d4/ : SUCCESS in 4m 24s\n- openstack-tox-molecule https://logs.opendev.org/41/674641/2/check/openstack-tox-molecule/640eae5/tox/reports.html : SUCCESS in 25m 50s\n- openstack-tox-docs https://logs.opendev.org/41/674641/2/check/openstack-tox-docs/b4b9363/ : FAILURE in 3m 57s\n- openstack-tox-lower-constraints https://logs.opendev.org/41/674641/2/check/openstack-tox-lower-constraints/7026206/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a77633cd73079b29f43f7238de2fa3376931d749","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-13 10:02:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87db70b6b7b929c1f13c78bd9f440d01d8655cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 12:54:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3a6d35a7e31745749a72b2687c68c479 : SUCCESS in 2h 41m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2e6e204a1b3241afa0a93a0addb4cb77 : SUCCESS in 2h 14m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/24fa5834dcb14f2295eeac81f46755ed : SUCCESS in 2h 12m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0c2906e903a045409f19f90e154e3b68 : SUCCESS in 2h 42m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e8486b78bc8345fb8f5bdb65a8d32644 : SUCCESS in 2h 31m 23s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/c92948adc11b4eacb67be2e922f1831c : FAILURE in 2h 22m 27s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b257d41bab6a472f8346acaab5dea0d7 : SUCCESS in 2h 19m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8c92d58e6947feb3726c61303e4ae4 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3981458c1994443c929cdae3d3051dd6 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf257dde3cc43178b6994c3fecb7980 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a089bdb1ae8a4844b2c727f4037aaeb1 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54607cc161734c9ea98aef2f584a0490 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64d554ab1eca44f590b36ebd2652baff : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"4dd6b70fdcc4319483a4930c7f8c590e68ec03dd","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-09-18 16:51:18.000000000","message":"Abandoned\n\nhttps://specs.openstack.org/openstack/tripleo-specs/specs/policy/patch-abandonment.html","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c61259501bd985bea5855abc84cddc576463d323","revisions":{"412bdd428d71e4ba7f41ab3361cfa58f3da9ae61":{"kind":"REWORK","_number":1,"created":"2019-08-05 15:58:45.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/41/674641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/41/674641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/1"}}},"commit":{"parents":[{"commit":"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 15:57:31.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 15:57:31.000000000","tz":120},"subject":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman","message":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman\n\nChange-Id: I4620a7e026bf1ba276d9219630972c14fa0a0137\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/412bdd428d71e4ba7f41ab3361cfa58f3da9ae61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/412bdd428d71e4ba7f41ab3361cfa58f3da9ae61"}]},"branch":"refs/heads/master"},"deb68b8ce1941f5931444eeb320f370c51fbb475":{"kind":"REWORK","_number":2,"created":"2019-08-05 16:00:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/41/674641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/41/674641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/2"}}},"commit":{"parents":[{"commit":"218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e","subject":"Add autodocumentation for roles and library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/218bef9cbe384a4d77b46b6f8b7cdd35387cfe3e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 15:57:31.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 16:00:18.000000000","tz":120},"subject":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman","message":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman\n\nChange-Id: I4620a7e026bf1ba276d9219630972c14fa0a0137\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/deb68b8ce1941f5931444eeb320f370c51fbb475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/deb68b8ce1941f5931444eeb320f370c51fbb475"}]},"branch":"refs/heads/master"},"c61259501bd985bea5855abc84cddc576463d323":{"kind":"REWORK","_number":3,"created":"2019-08-13 10:02:03.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/41/674641/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/41/674641/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/41/674641/3"}}},"commit":{"parents":[{"commit":"e3eb813fee773eccee7e5155a22c141df8b93c0f","subject":"Remove useless meta directory for the Ansible ntp role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e3eb813fee773eccee7e5155a22c141df8b93c0f"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-05 15:57:31.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-13 10:01:35.000000000","tz":120},"subject":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman","message":"[WIP] Migrate containerized-undercloud-docker to containerized-undercloud-podman\n\nChange-Id: I4620a7e026bf1ba276d9219630972c14fa0a0137\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c61259501bd985bea5855abc84cddc576463d323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c61259501bd985bea5855abc84cddc576463d323"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
