)]}'
{"id":"openstack%2Ftripleo-validations~674771","triplet_id":"openstack%2Ftripleo-validations~master~I14bdbbea58dce26993c98f959320dbf063406268","project":"openstack/tripleo-validations","branch":"master","hashtags":[],"change_id":"I14bdbbea58dce26993c98f959320dbf063406268","subject":"Revert \"Improve Ceph health checks to work with custom cluster names\"","status":"MERGED","created":"2019-08-06 10:36:25.000000000","updated":"2019-08-06 20:22:00.000000000","submitted":"2019-08-06 20:22:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674771-1565122920645-df5dc711","meta_rev_id":"6f864c0e0d7e93294ca0b88024da6fd1f61fbf36","_number":674771,"virtual_id_number":674771,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-08-06 20:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2019-08-06 12:59:30.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":2,"date":"2019-08-06 11:12:06.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-08-06 13:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-06 11:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":1,"date":"2019-08-06 11:12:32.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 10:36:48.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-06 10:39:40.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-06 10:39:59.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2019-08-06 12:59:30.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-08-06 13:34:56.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-06 20:22:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a33c5a8cafbf32458906805daf6d42062adc867","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-06 10:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99787d341e6041ededc8b62b2a258528456daac9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 10:36:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78d789a580173e48bc11656cc89394ddaaeef20a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-06 10:39:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d89859ebd4d377b91fbfad4a31f471b0d881f2b0","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-06 10:39:59.000000000","message":"Patch Set 1:\n\nWould have been good to add #1838556 as related bug.","accounts_in_message":[],"_revision_number":1},{"id":"774a7572cd389df727e7b61f5f82786e29aad6a1","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-06 11:12:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"91f97b27d79aa21c68fc29fe92505899fecd9105","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-06 11:12:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"302261695ee47c054cf33e6cf0ffecc5c722e322","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-08-06 12:59:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"32e6a05f5ff610da3f72061faa109e2cc793ecdc","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-06 13:34:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3e602a180a6291b806153ff453222ac531d79813","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-containers-multinode/fa13d18/ : SUCCESS in 2h 24m 03s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-undercloud-containers/3cd7e39/ : SUCCESS in 2h 27m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7bc9bfd/ : SUCCESS in 2h 11m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/18b0f44/ : SUCCESS in 2h 17m 11s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-standalone/134116a/ : SUCCESS in 1h 54m 55s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/71/674771/2/check/tripleo-ci-fedora-28-standalone/a8f71f5/ : SUCCESS in 1h 58m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/71/674771/2/check/tripleo-ci-centos-7-standalone-upgrade/9abc32f/ : SUCCESS in 2h 01m 01s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/71/674771/2/check/openstack-tox-pep8/58ebc55/ : SUCCESS in 5m 40s\n- openstack-tox-py27 https://logs.opendev.org/71/674771/2/check/openstack-tox-py27/14c54f9/ : SUCCESS in 5m 04s\n- openstack-tox-py36 https://logs.opendev.org/71/674771/2/check/openstack-tox-py36/53920c3/ : SUCCESS in 5m 24s\n- openstack-tox-py37 https://logs.opendev.org/71/674771/2/check/openstack-tox-py37/c8a3d64/ : SUCCESS in 5m 47s\n- openstack-tox-molecule https://logs.opendev.org/71/674771/2/check/openstack-tox-molecule/cf9dce4/tox/reports.html : SUCCESS in 29m 14s\n- openstack-tox-docs https://logs.opendev.org/71/674771/2/check/openstack-tox-docs/fb7e672/html/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://logs.opendev.org/71/674771/2/check/openstack-tox-lower-constraints/d49bd7f/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"80cdd90fdb3ad46f3eb7634af35df1c35358800e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:07:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18f3d1d6719fe79c66b982a8e3e988eb4b3f46d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:22:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/71/674771/2/gate/tripleo-ci-centos-7-containers-multinode/b401db8/ : SUCCESS in 2h 41m 57s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/71/674771/2/gate/tripleo-ci-centos-7-standalone/2bcb01f/ : SUCCESS in 1h 50m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/71/674771/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1156fa1/ : SUCCESS in 2h 35m 01s\n- openstack-tox-pep8 https://logs.opendev.org/71/674771/2/gate/openstack-tox-pep8/87ddb65/ : SUCCESS in 4m 51s\n- openstack-tox-py27 https://logs.opendev.org/71/674771/2/gate/openstack-tox-py27/4bb705e/ : SUCCESS in 6m 22s\n- openstack-tox-py36 https://logs.opendev.org/71/674771/2/gate/openstack-tox-py36/7554768/ : SUCCESS in 5m 50s\n- openstack-tox-py37 https://logs.opendev.org/71/674771/2/gate/openstack-tox-py37/cb3d2f6/ : SUCCESS in 5m 19s\n- openstack-tox-molecule https://logs.opendev.org/71/674771/2/gate/openstack-tox-molecule/65f6465/tox/reports.html : SUCCESS in 29m 23s\n- openstack-tox-docs https://logs.opendev.org/71/674771/2/gate/openstack-tox-docs/fe47102/html/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/71/674771/2/gate/openstack-tox-lower-constraints/c4f3185/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6f864c0e0d7e93294ca0b88024da6fd1f61fbf36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:22:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e19638c93e352970b27ae556faac70baabac73ea","revisions":{"db4d4726d9f6a632df27426c73c093bb4271e9e8":{"kind":"REWORK","_number":1,"created":"2019-08-06 10:36:25.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/71/674771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/71/674771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/1"}}},"commit":{"parents":[{"commit":"ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344","subject":"Improve Ceph health checks to work with custom cluster names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-06 10:36:25.000000000","tz":0},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-06 10:36:25.000000000","tz":0},"subject":"Revert \"Improve Ceph health checks to work with custom cluster names\"","message":"Revert \"Improve Ceph health checks to work with custom cluster names\"\n\nBreaking validations\n\nThis reverts commit ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344.\n\nChange-Id: I14bdbbea58dce26993c98f959320dbf063406268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/db4d4726d9f6a632df27426c73c093bb4271e9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/db4d4726d9f6a632df27426c73c093bb4271e9e8"}]},"branch":"refs/heads/master"},"e19638c93e352970b27ae556faac70baabac73ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-06 11:12:06.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/71/674771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/71/674771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/71/674771/2"}}},"commit":{"parents":[{"commit":"dcf051ec42071974ba75276c20366f57d4d045e9","subject":"Merge \"Improve Ceph health checks to work with custom cluster names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/dcf051ec42071974ba75276c20366f57d4d045e9"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-06 10:36:25.000000000","tz":0},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-08-06 10:56:34.000000000","tz":120},"subject":"Revert \"Improve Ceph health checks to work with custom cluster names\"","message":"Revert \"Improve Ceph health checks to work with custom cluster names\"\n\nBreaking validations\n\nThis reverts commit ff4d7ceec5a24e0b53d0b2effc2e7d23cb895344.\n\nChange-Id: I14bdbbea58dce26993c98f959320dbf063406268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e19638c93e352970b27ae556faac70baabac73ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e19638c93e352970b27ae556faac70baabac73ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
