)]}'
{"id":"openstack%2Ftripleo-validations~675616","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I66b796fab1d8651163226febbc4e99648a9ecc6a","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"bp/validation-framework-infra-backport","hashtags":[],"change_id":"I66b796fab1d8651163226febbc4e99648a9ecc6a","subject":"Linting hardening with pre-commit","status":"MERGED","created":"2019-08-09 12:46:30.000000000","updated":"2019-08-19 17:39:16.000000000","submitted":"2019-08-19 17:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675616-1566236356483-8c476a95","meta_rev_id":"6d48580a541c0f9310dda2e1edbc3e1a0aef6ce7","_number":675616,"virtual_id_number":675616,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-19 17:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-08-19 08:58:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-08-19 14:14:58.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 14:14:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-19 17:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ac71dc156bb60eba859bed7a3e1584fa9305720","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-09 12:46:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"009fea8063f92b83a393825846dc9113cd5f59c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 19:30:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/16/675616/1/check/tripleo-ci-centos-7-containers-multinode/d57a217/ : SUCCESS in 2h 09m 59s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/16/675616/1/check/tripleo-ci-centos-7-undercloud-containers/68da204/ : SUCCESS in 1h 34m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/16/675616/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4632c77/ : SUCCESS in 2h 10m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/16/675616/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8394ead/ : SUCCESS in 2h 43m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/16/675616/1/check/tripleo-ci-centos-7-standalone/d068fe4/ : SUCCESS in 2h 04m 41s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/16/675616/1/check/tripleo-ci-fedora-28-standalone/22b656b/ : SUCCESS in 1h 47m 26s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/16/675616/1/check/openstack-tox-pep8/c3c28f6/ : SUCCESS in 4m 24s\n- openstack-tox-py27 https://logs.opendev.org/16/675616/1/check/openstack-tox-py27/40c39ae/ : SUCCESS in 4m 57s\n- openstack-tox-py35 https://logs.opendev.org/16/675616/1/check/openstack-tox-py35/a811687/ : SUCCESS in 6m 24s\n- openstack-tox-py36 https://logs.opendev.org/16/675616/1/check/openstack-tox-py36/180812f/ : SUCCESS in 5m 24s\n- requirements-check https://logs.opendev.org/16/675616/1/check/requirements-check/d27ed1a/ : SUCCESS in 3m 34s\n- openstack-tox-docs https://logs.opendev.org/16/675616/1/check/openstack-tox-docs/7efa553/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://logs.opendev.org/16/675616/1/check/build-openstack-releasenotes/7efb0a1/html/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/675616/1/check/openstack-tox-lower-constraints/8a9df2d/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"df96b37501dd33d64461d3ea053e8a6bada9108a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-12 11:33:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"643126320f5143a5e7a4e9eba4f67fdfcc953b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:13:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/16/675616/2/check/tripleo-ci-centos-7-containers-multinode/5ddf36f/ : SUCCESS in 2h 08m 18s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/16/675616/2/check/tripleo-ci-centos-7-undercloud-containers/9e422b3/ : SUCCESS in 1h 38m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/16/675616/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/769ed87/ : SUCCESS in 2h 23m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/16/675616/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/56c32c0/ : SUCCESS in 2h 35m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/16/675616/2/check/tripleo-ci-centos-7-standalone/0b1a8cf/ : SUCCESS in 1h 48m 22s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/16/675616/2/check/tripleo-ci-fedora-28-standalone/d4ebaf3/ : SUCCESS in 2h 05m 23s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/16/675616/2/check/openstack-tox-pep8/debc59f/ : SUCCESS in 4m 34s\n- openstack-tox-py27 https://logs.opendev.org/16/675616/2/check/openstack-tox-py27/6776a42/ : SUCCESS in 3m 59s\n- openstack-tox-py35 https://logs.opendev.org/16/675616/2/check/openstack-tox-py35/56e6b71/ : SUCCESS in 5m 37s\n- openstack-tox-py36 https://logs.opendev.org/16/675616/2/check/openstack-tox-py36/753d222/ : SUCCESS in 3m 41s\n- requirements-check https://logs.opendev.org/16/675616/2/check/requirements-check/7d3f831/ : SUCCESS in 2m 57s\n- openstack-tox-docs https://logs.opendev.org/16/675616/2/check/openstack-tox-docs/7b2c069/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://logs.opendev.org/16/675616/2/check/build-openstack-releasenotes/263a9a4/html/ : SUCCESS in 2m 53s\n- openstack-tox-linters https://logs.opendev.org/16/675616/2/check/openstack-tox-linters/7f64a00/ : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/675616/2/check/openstack-tox-lower-constraints/90de1ec/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b914013e2f9c14a41493742a5b3d8abad5689cab","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-13 09:21:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"71ce91be0d575e61633cc1acf38df846af4ab619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 12:07:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3a2374a11d264c9abe264fa659d1d249 : SUCCESS in 2h 09m 37s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3a3269de981742f5b278d1924c3fd438 : SUCCESS in 1h 41m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5b62d72bc8a04fc89ab406af4f798098 : SUCCESS in 2h 27m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c31bc22bb29244e7a430a18ea8aa66c0 : POST_FAILURE in 2h 34m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8363261581dd42e3b263850f30013eb0 : SUCCESS in 2h 14m 23s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/fdc65bfc23994b05833c1ebefe4fb712 : SUCCESS in 1h 59m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f823b4cc4cb4dc1832d5fafe8a9fb86 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15c3b202ad134919afd6f41de11b237f : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08ce11bdcc0a44b09c34c0a19c186f39 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6ac589e84f45f7a243953aa95a849c : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/17a2c33c57544ad4a3b7179cdbdda3a3 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972873a536324cb38dea30186c697efb : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b158628c2e2546629e644b808967126d : SUCCESS in 4m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6f55aa86bd8a49ffb542da980d800e48 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/291b9e946368430faece76b22cd98693 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"99fb7adf08aae28fc800067196d2ca18080cd383","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-16 22:15:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51548cba85f7f6612d6ac5e366ea6051cf825b75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:15:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a7d85b3856deba4b894937bea7b6b9e07968bb8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 00:30:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e7b194e55bc4764b3ffd0fcfa092f8a : SUCCESS in 2h 05m 11s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/89c20cd99aa04bee990fa5d94214453a : SUCCESS in 2h 13m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a6733e070894cd88363538db268b72a : FAILURE in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0377fddd0b04dda9202e7dddf357dae : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8393ba56ae864dbcafa11f05b4e4f5be : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a3ea41700274aa1bac9bb317355b062 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40d5b5dd2ef2445a8d5c8ce43a78e1e4 : SUCCESS in 4m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2086eff60ec4b26a24a507976fb90b8 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ce163a4c854102bfa3c0225e9b4da9 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e367b6cb9504f859fff3b3bdbc6a6c7 : SUCCESS in 3m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/284322877dfb4fd4a7174ea46febbfa4 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9205d1cc29264707b0fc8424663f3905 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ce5e407ae003195d2261ab231a791d585cd11568","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:43:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"daa15cb1f09125a9fc6ed95862c52a8f5f20376b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:58:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"7903fd68692aabf3399980e4cda67a06e12a7dc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 12:17:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4ece2bf123f94fb794997d9543095d77 : SUCCESS in 2h 46m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/76b224a1a3bf4333b174017ee7968b19 : SUCCESS in 2h 06m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/22df0711b9454c08b97e3d24740f58e0 : SUCCESS in 2h 49m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0277c6b839ce4f169913869b2c58f811 : SUCCESS in 3h 14m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/60214356591d4221822e919644ddf3b0 : SUCCESS in 2h 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2cd2e29e4248709bda8dea6c1a4e4d : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52702d2e46474c099bea9b9c82455fa5 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/353a6ad06da34da092ae3905d92803e4 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4098660d447b416382f32214e1a96a97 : SUCCESS in 6m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4125fa474d314e02a6e99f037089c6b7 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2012a869a154cd3a86dc23fa02ff087 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16be4cb4930b492182d8df78202e9aad : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2843488cde934503a6506bc7dc3adcb8 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ee188e2e2ac4748a5ab354210ecc281 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"cd92145fc909a3e6665cd8cd6cffb20a8770c638","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 14:14:58.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7ff4cb872f1b9b6708edd490598e432ec1369481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:15:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a0767875b821ee2922cccb5fee851f928a98bca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:39:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/01b396a53f874d1c97921b3e5c789500 : SUCCESS in 2h 29m 25s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6870cd6494ff49e19c50ed213037b15b : SUCCESS in 1h 57m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bc11cb07417d402795469848c117b2e8 : SUCCESS in 2h 39m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b44fd33d3a48b0be163b193ff870df : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/073ce9dcca714d7c88ee3a28068699c3 : SUCCESS in 6m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40d306a141e54910baa744452ffd628a : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be54a08c7ac34910ad9910721bffbb9d : SUCCESS in 7m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6fe7905ee81f4284b90d6cfb461acd03 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe29d46eb614c47a420fb3fe979c1b6 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ee4b9c8be1488bbb41b83d39e0ffbf : SUCCESS in 4m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e438f19859a940f49cd652fd404a550f : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99c990978d2e4d33b3aafb723e51e456 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"72a20f7170b3c33879df4a91cabf5c6ec86c6bf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:39:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4d309d8e542d81e15e6ce669791a7ded5970182e","revisions":{"e0c9da66f7e0bb32aa6305f88b4d719d67ba85f9":{"kind":"REWORK","_number":1,"created":"2019-08-09 12:46:30.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/16/675616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/16/675616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/1"}}},"commit":{"parents":[{"commit":"80a9f7b99bcdff3aab792c571683800f0dd7c8ad","subject":"Simplifying the check of the latest package version validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/80a9f7b99bcdff3aab792c571683800f0dd7c8ad"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-09 12:45:58.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit cac3e7101d89c743ce2593141b9da4735be2fd33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0c9da66f7e0bb32aa6305f88b4d719d67ba85f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0c9da66f7e0bb32aa6305f88b4d719d67ba85f9"}]},"branch":"refs/heads/stable/stein"},"581a75748b2d5b470361c8669273d2842bdee2f8":{"kind":"REWORK","_number":2,"created":"2019-08-12 11:33:07.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/16/675616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/16/675616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/2"}}},"commit":{"parents":[{"commit":"80a9f7b99bcdff3aab792c571683800f0dd7c8ad","subject":"Simplifying the check of the latest package version validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/80a9f7b99bcdff3aab792c571683800f0dd7c8ad"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-12 11:32:22.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit cac3e7101d89c743ce2593141b9da4735be2fd33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/581a75748b2d5b470361c8669273d2842bdee2f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/581a75748b2d5b470361c8669273d2842bdee2f8"}]},"branch":"refs/heads/stable/stein"},"1875104ab11032165ecac7c82457a709d5f5c1eb":{"kind":"REWORK","_number":3,"created":"2019-08-13 09:21:50.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/16/675616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/16/675616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/3"}}},"commit":{"parents":[{"commit":"d3c7f735dd8534df8df1e6b177a9a850f9e510c6","subject":"Merge \"Use custom Ceph cluster name in validations\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d3c7f735dd8534df8df1e6b177a9a850f9e510c6"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-13 09:21:39.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit cac3e7101d89c743ce2593141b9da4735be2fd33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1875104ab11032165ecac7c82457a709d5f5c1eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1875104ab11032165ecac7c82457a709d5f5c1eb"}]},"branch":"refs/heads/stable/stein"},"4d309d8e542d81e15e6ce669791a7ded5970182e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-19 08:58:09.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/16/675616/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/16/675616/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/16/675616/4"}}},"commit":{"parents":[{"commit":"6f827f5ff81c243ca1d67b86603327736a4058c6","subject":"Merge \"Correct \"repos\" validation and adds Molecule tests\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6f827f5ff81c243ca1d67b86603327736a4058c6"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-30 09:35:05.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-19 08:49:54.000000000","tz":120},"subject":"Linting hardening with pre-commit","message":"Linting hardening with pre-commit\n\n- Adds bashate test using v0.6.0\n- Bump pre-commit-hooks release to v2.2.4\n- Bump ansible-lint release to v4.1.0a0\n- Fix some minor flake8 errors\n\nChange-Id: I66b796fab1d8651163226febbc4e99648a9ecc6a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit cac3e7101d89c743ce2593141b9da4735be2fd33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4d309d8e542d81e15e6ce669791a7ded5970182e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4d309d8e542d81e15e6ce669791a7ded5970182e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
