)]}'
{"id":"openstack%2Ftripleo-validations~676112","triplet_id":"openstack%2Ftripleo-validations~master~I6e09e0c400df5a7bf35b524cd576ae17c931ba9c","project":"openstack/tripleo-validations","branch":"master","hashtags":[],"change_id":"I6e09e0c400df5a7bf35b524cd576ae17c931ba9c","subject":"Removed older version of python","status":"MERGED","created":"2019-08-13 06:30:11.000000000","updated":"2019-08-19 15:35:12.000000000","submitted":"2019-08-19 15:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676112-1566228912764-0e67f363","meta_rev_id":"12b04a50aa035378cc1b951e8fb144cb50bbfb15","_number":676112,"virtual_id_number":676112,"owner":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-19 15:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2019-08-19 13:03:57.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-08-16 22:14:50.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2019-08-19 13:03:57.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 22:14:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-19 13:03:57.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-19 15:35:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5616ca732155c364da967c37afe4a53c468d64f6","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-08-13 06:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f844ef9450385d5f5ef88e1853052bd7f51ea3d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 08:53:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/370d1c06a5a7464c83bc41015b8079f0 : SUCCESS in 2m 27s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/27fe91f1aea743cdb9236132cc67bba9 : SUCCESS in 1h 53m 58s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/6fe65d230add4e7e9b07416ed739c6c7 : SUCCESS in 1h 49m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/dcce77d08acf45ceb362515934f689e4 : SUCCESS in 2h 17m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819ce277f2d74ec69ac18a110134967b : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/667f8c9237394b809b2a193a93b58053 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3242a9debd4419a7b3c6c8749c9867 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b08e26a28d4414898d848e64b38db64 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75efb853c3884ada98f45ac9bf84027f : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8449da68b1f5421aad153e497c9e7d41 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c9e06299411978f06861d099e253baa6f5ac1286","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-16 22:14:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a1fce857cd6cfc96d0b405a4a1002ded53294ac","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:43:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f53bda46c3134ac42b606a58bb43fbc961019d94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:43:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7902718468abd3e95f0285c4ed6bcaff37d13851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 10:06:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9b477e5a915c4f1ca894e82ca552c603 : SUCCESS in 3m 01s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c52b5fa2f3cc4d42a7e4e9af6f334804 : FAILURE in 1h 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de9debeeceb4b7ba8fc599b0ce45c61 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63ec872ab32844acbe20b3c751565a3f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/645b496203c141fa8430747517ae3f67 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b962ac1a0b6424bad02c3ef8cb40d21 : SUCCESS in 5m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf2ca17942d04516851c3d27ec8620ba : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240fff7ab7f3443fbd0882df98ecf9f6 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"56451997dd3f5b2f35a03066c1d40ba775939734","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 10:50:36.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b73e417adf616daabec71ca55e919f1b800f859d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 13:03:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4616891d7ac1978b0b1732661ca8c9fb83b28547","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:25:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8051e218ca3342b88f22141adc93816a : SUCCESS in 2m 52s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d57d55279f4d4b16a05825e8b2cec9fe : SUCCESS in 1h 42m 20s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd2501c94cdb45288c893f35bb634842 : SUCCESS in 2h 30m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfe685876d24715af54829e5f5de68f : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e57790b5f394fe5960aef587088e735 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/659e43660c0b4826847837ee71f9d1d4 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08d536b6f3eb4060bdef722b5f776156 : SUCCESS in 4m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5596adafc9e246bdb315ad830003606e : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26d69a3fb1e24168977b9a14d78f0569 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9c0511a274f87ac5c14c1d9174690bb0880aef9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:25:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0eea0da6b37c21c09059ff40e9cf2c34bdb089ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:35:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"12b04a50aa035378cc1b951e8fb144cb50bbfb15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:35:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63461bb613a047cc8feee115bfd87be7 : SUCCESS in 3m 27s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/34380129c3c94c7ea5c13e34b2ca2401 : SUCCESS in 2h 03m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77350303d9cf48a9bfa23c2888145070 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c685fdb6f4414e1c85888abcf5ae0bf8 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0202ecbc657648a09daf2a12df1d9ccd : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b98bfaf0f734ed7a5515c67f5a2ad31 : SUCCESS in 5m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/761de0c916f74f5790d196688425ae82 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/410f86d84e8c4574b59d326fff32d991 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2ac1730d170b8e63ab339be195e04fba892ff8c","revisions":{"b2ac1730d170b8e63ab339be195e04fba892ff8c":{"kind":"REWORK","_number":1,"created":"2019-08-13 06:30:11.000000000","uploader":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"ref":"refs/changes/12/676112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/12/676112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/676112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/676112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/12/676112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/12/676112/1"}}},"commit":{"parents":[{"commit":"e3eb813fee773eccee7e5155a22c141df8b93c0f","subject":"Remove useless meta directory for the Ansible ntp role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e3eb813fee773eccee7e5155a22c141df8b93c0f"}]}],"author":{"name":"Vieri","email":"15050873171@163.com","date":"2019-08-13 06:28:13.000000000","tz":0},"committer":{"name":"Vieri","email":"15050873171@163.com","date":"2019-08-13 06:30:09.000000000","tz":0},"subject":"Removed older version of python","message":"Removed older version of python\n\nChange-Id: I6e09e0c400df5a7bf35b524cd576ae17c931ba9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b2ac1730d170b8e63ab339be195e04fba892ff8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b2ac1730d170b8e63ab339be195e04fba892ff8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
