)]}'
{"id":"openstack%2Ftripleo-validations~705429","triplet_id":"openstack%2Ftripleo-validations~master~I9016497305b4ec1a1e49b4b1c3cb23d219a42458","project":"openstack/tripleo-validations","branch":"master","topic":"service_status","hashtags":[],"change_id":"I9016497305b4ec1a1e49b4b1c3cb23d219a42458","subject":"Updated/corrected service_status","status":"MERGED","created":"2020-02-03 12:20:30.000000000","updated":"2020-07-30 07:28:31.000000000","submitted":"2020-02-24 19:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705429-1582573376945-5004bf0a","meta_rev_id":"88210309060a9f91e65281956da8668df6dc7306","_number":705429,"virtual_id_number":705429,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 19:42:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-24 13:01:47.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2020-02-24 09:40:32.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2020-02-19 16:03:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2020-02-24 09:40:32.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 17:46:30.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-02-19 16:03:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-24 09:40:32.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-02-24 19:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e70c3a7d9865fcd4a6f9fb1d9bf3947af441e97","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-03 12:20:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"420d1a9cc642dc5d8d5a720c5671368da41b13ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 16:08:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f1faf717f3584f079afd4be2cc39268f : SUCCESS in 3h 15m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/94eba517772246428e85ecce8808ce23 : SUCCESS in 2h 08m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0564cc0300ae4477ab84865ecdcfd57d : SUCCESS in 2h 04m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8dab5c252aa14c09a3666ad3df118d13 : SUCCESS in 2h 24m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d5d22f2872e841458e68724841edc69a : SUCCESS in 1h 52m 03s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d5ce93b596d844fd9b473e5ac9121f90 : TIMED_OUT in 3h 17m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7248107d522e4aa59f8cd3cd15eadeac : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c45f3522da44401b3e628f38c584ce0 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb38cc794b9848b7bbaecc4ae54cf418 : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17a1ad1334b24984a3e3a9e916340f7b : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37b9f1ac46894e77a90791678690aaf8 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e74aeafa2c9a432690bfe5ddfe9bcd55 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"13ac8cdfadca9d2d15708881ec6553546db70910","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-02-03 17:46:30.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLGTM! Just a minor comment about one of the tasks.","accounts_in_message":[],"_revision_number":1},{"id":"8a515872a8ea1f0efdc715fe2aa4f4531812ea6f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-07 23:41:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"342e6f2d5c82c8b07b6d7dcba4d1d36f3e4ad87e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-10 08:56:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06b27794ddf03216bc9b8b5f683cba28d3236f44","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-10 09:08:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2501c9716aa18d0dda026fb787dfd6fd222d6292","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-10 09:09:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93006d43be3ce495c0113cca324774a474812ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:54:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b3d9a1afdff34d879008fa7c62f74380 : SUCCESS in 2h 40m 15s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/46b43d188edf4397a1b6295ce688a4f4 : SUCCESS in 2h 00m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/aafb70b0286548f086e384662f9e6648 : SUCCESS in 2h 23m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/498d0ae47ce04dd093c2c433f7a12ca1 : SUCCESS in 2h 13m 23s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cb1e4aa5689c4e38b6c7fa0dba54b962 : SUCCESS in 2h 00m 04s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a36408b6c19946cdb6a61ce8b1308742 : SUCCESS in 2h 13m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052f9a96df1c450fa883150d64c7adb0 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40e61ffffb834ecc90cd39690cac958e : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec191e9dc28441868477d66fd3f21d39 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0acc7ebe91f94af8a74c41cea83b6cbc : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b62fde9e8e5425fb31e98b7a20e5f0a : FAILURE in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8aa395eb41c4c71968b082028c94226 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b45ab6081b10c35a02ec711e6f9e47fb0ea0c97e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-10 12:43:15.000000000","message":"Patch Set 3: Workflow-1\n\nWill take the opportunity to add molecule testing on that role. Rebasing on https://review.opendev.org/704158 + some more dev.","accounts_in_message":[],"_revision_number":3},{"id":"56bf2c98a6e31966e148e6fa711aff001c46c4ac","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-11 15:02:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d1a15fda8c77cce266157d7df0009451d727859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:03:09.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"adee481ed81b70563fe7702d926db0d397836cd6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-11 15:06:11.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"eba82c5e34fc014ec5ef055fdd72100fb35c73f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:26:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ab233c9e6c49413c8fa5fe2627ca3bbf : FAILURE in 10m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8286c529468e4271be2cea8080003f46 : FAILURE in 10m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ae19d14a09114a0a8f9599f8596c1e7e : FAILURE in 12m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9873c7c55c0a4ce4a983e05142f36fce : FAILURE in 11m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b2b2e1c490094e92b2544317cc341ef3 : FAILURE in 12m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/15f2ea21bebe41528292fe9cc57a47ec : FAILURE in 14m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf1feb6e078444c875d6a4e3d54ef2e : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c77cacbca3f142d7871d832ac7407f5b : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18666fafaed94721b8acf2b3f3977744 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1950fa44b3954022aab7b616f7045f05 : SUCCESS in 7m 42s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/0b05189208204d7fa26e4704ea4abbd9 : FAILURE in 8m 26s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c584db63eae54c1c8079a2b43a3e351c : FAILURE in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b14fc3cc1754cd7a1e81d6b743a0f7f : FAILURE in 6m 26s","accounts_in_message":[],"_revision_number":5},{"id":"531cdd02eb5803fca717641cae1ccf02a5326099","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-12 08:13:47.000000000","message":"Patch Set 5:\n\nrcheck","accounts_in_message":[],"_revision_number":5},{"id":"db7c9c77861e78c5024e43d066c6162f5e359b13","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-12 08:13:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8b72a4671016dd1298210518e4015dcae444abe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 11:14:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d30da52f2d7f492887c791be84e6dcfe : SUCCESS in 2h 26m 31s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2db689c8db564209b12d072b3f5d0314 : SUCCESS in 2h 07m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c8472535a71248cfa77e98e83d0e4f83 : SUCCESS in 2h 31m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ca8b9cce393940488353be4624af970e : SUCCESS in 2h 43m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d38248f3b9d84b90bd11dfbec94a6cfd : SUCCESS in 1h 42m 35s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f51650e951934953ad2002a914e1f60e : SUCCESS in 2h 26m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f70eb6c7de49638dca1752c4bdacec : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4adf4199847b4195a7e11c3ad042aad1 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f988f359c234507961e8346253a0ee5 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fbff188689b43a39ff1cc65268594b1 : SUCCESS in 7m 12s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/2ee8e2dbb35a4380825ab7bd841214ff : FAILURE in 10m 13s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1113079065c84942ac92b2af6bcb3479 : FAILURE in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddae35f67a0c4512a865a25a94d6bda8 : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":5},{"id":"b050c87394f1743ad24023bafd68be8ac48c8111","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-14 12:01:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3dad96ebfdd6eeca8750a804dfeb7f6580eeca70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:52:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/88e22ca10fca4b15a9bd8f57f49e3361 : SUCCESS in 2h 46m 04s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/671991a1a69a4ed9afdda4b758547c5a : SUCCESS in 1h 49m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/882070b94d194ae89d019db85a1c8fec : SUCCESS in 2h 26m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6d575e64d90f4448a5af283162d8ce96 : SUCCESS in 2h 29m 29s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/bf7c29dc63b34f0596feab9da2c9d51f : SUCCESS in 1h 55m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/4a3d3e6b5b2441f0878f9b28ffe130d0 : SUCCESS in 2h 11m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93f6b99007a44eab056ff22052f7f66 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6330415789458faaa15812e96f1aee : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6afc68f79ecd478b8aa8f5d4d65d4661 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/141b1b1934c84258806fe107a975f04c : SUCCESS in 6m 01s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/abc305fc56034d08ac8faace3c7a23f7 : FAILURE in 9m 32s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9394ef4580b546998107c19fe56c9772 : FAILURE in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a615badd2bd4477d94ce1db204b3d55d : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5eab9cf95435859807133739a854f713cbecbb41","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-18 12:27:05.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2dc44636a449b9d03c5982a28a602a2d121af5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 15:30:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/913155948020488ca0813c27970ec71c : SUCCESS in 2h 44m 59s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/80a0b39861de420b8b18c2f6bdf157a0 : SUCCESS in 2h 06m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/534d9c78b512463e8dbee2a1748f05ae : SUCCESS in 2h 24m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bf410e1a3c0b47e199b8bfcbd0162976 : SUCCESS in 2h 22m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b65f791bb97f4cc29e8bb36735dd2418 : SUCCESS in 1h 55m 51s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f7c1b0bc99f3432fa011b1aa54f110c4 : SUCCESS in 2h 07m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4aa024000b944f59eaac21323e92e63 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ea5cfc4bf5943968120fa2c9ae14bc7 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62412f43f0c14db8a0a6892bbd441ca7 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746dff684f07489d8bea73622977b27d : FAILURE in 6m 46s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/b1604a8fb53842118557dd1c3ffa5721 : FAILURE in 10m 08s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b39241fe316a4a0aa37bb935d8027d9f : FAILURE in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3369ae9939ab48939a3386b52b5809e6 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":5},{"id":"41851e94b180d87d9422e4c0ac701b0cd46a67ce","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-19 08:03:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fd920f0d21e86f1f4647efcf448442ba7df34261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 10:50:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6b86b707838d429fb1bc3543b570fb17 : SUCCESS in 2h 43m 36s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3bc0433c609045ad840966cb8f7fe26b : SUCCESS in 2h 03m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/71780b70376540fb8a4d4540ed4bdafd : SUCCESS in 2h 25m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2e29c3b03bcd45889f14b9f2affa4a48 : SUCCESS in 2h 31m 38s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1a3eac4b7a5f453483e017070efd5c57 : SUCCESS in 1h 51m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6680f516988347e1946f77b263820ad9 : SUCCESS in 2h 22m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af014b47f98b4530a3d2eca0ce78a1a2 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b56f3f095da49d5a6766759678bb08c : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc64172e933a4b798087250ea8a255ce : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d30d5809c6e84c68b59a26c3e18495f2 : FAILURE in 6m 57s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/4a45c2ecfa334018854f142bbdd713c0 : FAILURE in 12m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6174524c4da045709f8af30516b31d5b : FAILURE in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf2342343af4999b326f95c9c940a40 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":6},{"id":"de89aa166c54456bd080bfd4ba22a6d1056097ab","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-19 12:33:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"644c3e75063ca82744f548254f58df610455d174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:57:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8e587d59f5fa4a97b297db5499375bd8 : SUCCESS in 2h 49m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2165d10915d349e5a3e9c8a593dae1bc : SUCCESS in 1h 59m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/37d9d822948746168189c566194c6b75 : SUCCESS in 2h 20m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/564108222a0e43cfb0012b10bf32b839 : SUCCESS in 2h 39m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c75e786997ca4aea86e29e8f0f65db64 : SUCCESS in 1h 48m 54s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7e73ad13e1514ae0ba8289b5e4b72acd : SUCCESS in 2h 43m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554a2132808a4729a1d08b53c5ffaa70 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b2a7fc71a5c41678364cd04fdaabbf4 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6ff790ef8a34a74bcb546d670e8a2a8 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0677e75d70e404f899beb7dccee7f72 : FAILURE in 7m 37s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/fcdf012270884452a22b10f7afb69818 : FAILURE in 9m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cb3dfe47e8df4cc8962bfdcd3b8b197b : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6900edfc84e243a18e48a63bc757dc2e : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":7},{"id":"138439c173cb98e01d50fb41d825ee25259eaa75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-19 16:03:12.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1471a6f50777884886055d81266c840882243777","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-02-24 09:40:32.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"73a3f238e0f3903aa8f154ea8024ed22528d0f33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 09:40:51.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"64645a8439794c24fb01eb7acb6f3200bd47206e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:30:34.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e1dd65e620dc4a339a1f5a5ed79b7cfe : SUCCESS in 2h 33m 03s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/52928249a7a145f6b92eece456145cc4 : SUCCESS in 1h 33m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8f8054edec3b4f048f665c142064aaf1 : SUCCESS in 2h 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59ba2c928bf403b908eaa7fe76c301c : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f29f6a66f5f42b8a4464df45eef6a26 : FAILURE in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5ac17c0c3f845ff965a7a16eee44b1e : FAILURE in 6m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c229fab306a403f8b95337f40092a5e : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76af1cedd7ef4af8a0ca63107b24e0dc : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":7},{"id":"bc95f3dcccebc6fefa0a54d3345e9dde4feefbc5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-02-24 13:01:47.000000000","message":"Patch Set 7:\n\nrecheck\nNow that LP#1864399 patch is merged, it should pass... crossing fingers!","accounts_in_message":[],"_revision_number":7},{"id":"ea7ad05ae70b31e406e26d0a2d8290a785822e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:03:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4311d761ea2d49f390fffc29d3b8d470 : SUCCESS in 2h 56m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/40ab2bf5f7fb41ffa109408b4ae1d5a7 : SUCCESS in 2h 01m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8111b07c11164c48b405df375863bff0 : SUCCESS in 2h 16m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6a02011b451644878ebdbe3cd6c1e054 : SUCCESS in 2h 36m 57s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d91b5b579f4747babf0dff555a3308a9 : SUCCESS in 1h 46m 53s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/448791e4151f4df6909c2de2a087c2ef : SUCCESS in 2h 41m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1eb1d1abcc4b6fa3e6b8b4fe2391c8 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/720ed7f6850449099f4389a7bc0cc3db : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb5f881944ac474d84904948edafb05b : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c1fef103aa490d80b2108ad626a9cc : SUCCESS in 6m 18s (non-voting)\n- tripleo-validations-centos-7-molecule-service_status https://zuul.opendev.org/t/openstack/build/d40e1a1b7f6340ac80f272bc79399344 : FAILURE in 9m 32s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82e0532db2b043df8be6e2f842654cae : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/674b47c2200740e897d186e909ef0d8a : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":7},{"id":"df5d40432458354fca1a7e12148399bb794009c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:03:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c851adeed9afef99453a1e5f5e50746e17d7e42f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"7ed21e2e0d5fb950ec843cbee21f31be4b1c4bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:42:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/16a2ceb08e274122a930cf1b44a9dd38 : SUCCESS in 2h 44m 29s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d3b00cd31972444db00f0e75a4593a77 : SUCCESS in 2h 30m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f390e20e69c6469783051dd232f9271c : SUCCESS in 2h 45m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0c43bc92b4486fb461ceeb7b23ff2a : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59b40a08d874b77a1df33c8e0f66b77 : SUCCESS in 9m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1290e48d073a4327850504be4a7b6d2c : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/baf1ff2cea8f45f9a83587a2150ddd79 : SUCCESS in 11m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/875efb7071104938b8d98441203cf8d1 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":7},{"id":"0e9b71592ec960182b1b6f8942c116ea7add4bb4","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-07-30 07:28:31.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit cf087407270e569d6830d0bccfdba2ddcc2f0a93","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b026524f8ff7098d9ac811124dc1744e9c4d7de2","revisions":{"63cb5a0059d30515d36e8605070308d25bae8c76":{"kind":"REWORK","_number":1,"created":"2020-02-03 12:20:30.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/1"}}},"commit":{"parents":[{"commit":"d8246e9af9a7986efc1e343a8fb76efcbc9e8e45","subject":"molecule-xfs_check_ftypefailure (non-voting), remove from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8246e9af9a7986efc1e343a8fb76efcbc9e8e45"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/63cb5a0059d30515d36e8605070308d25bae8c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/63cb5a0059d30515d36e8605070308d25bae8c76"}]},"branch":"refs/heads/master"},"cdb5cedd4d780bcd05b539c632588f1f9397bb61":{"kind":"REWORK","_number":2,"created":"2020-02-10 09:08:21.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/2"}}},"commit":{"parents":[{"commit":"d8246e9af9a7986efc1e343a8fb76efcbc9e8e45","subject":"molecule-xfs_check_ftypefailure (non-voting), remove from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8246e9af9a7986efc1e343a8fb76efcbc9e8e45"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-10 09:08:02.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cdb5cedd4d780bcd05b539c632588f1f9397bb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cdb5cedd4d780bcd05b539c632588f1f9397bb61"}]},"branch":"refs/heads/master"},"04cf323415f5473f6498fcad9f317577a84e82c0":{"kind":"REWORK","_number":3,"created":"2020-02-10 09:09:50.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/3"}}},"commit":{"parents":[{"commit":"d8246e9af9a7986efc1e343a8fb76efcbc9e8e45","subject":"molecule-xfs_check_ftypefailure (non-voting), remove from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d8246e9af9a7986efc1e343a8fb76efcbc9e8e45"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-10 09:09:44.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/04cf323415f5473f6498fcad9f317577a84e82c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/04cf323415f5473f6498fcad9f317577a84e82c0"}]},"branch":"refs/heads/master"},"dd43c75dc2f371b8118dcc6ef9fbfb1aec600921":{"kind":"REWORK","_number":4,"created":"2020-02-11 15:02:54.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/4"}}},"commit":{"parents":[{"commit":"86e1e557bd6bc53260e66f9291cbcf00a970e2ea","subject":"Add basic molecule tests for uncovered roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/86e1e557bd6bc53260e66f9291cbcf00a970e2ea"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-11 15:02:40.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\nThis patch also adds some fancy molecule testing.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/dd43c75dc2f371b8118dcc6ef9fbfb1aec600921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/dd43c75dc2f371b8118dcc6ef9fbfb1aec600921"}]},"branch":"refs/heads/master"},"1e8d4e79602284f6272bff429a993796e257e6b9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-11 15:06:11.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/5"}}},"commit":{"parents":[{"commit":"6e7cd9cebf073e63b840a5bc355f09318bd6c705","subject":"Add basic molecule tests for uncovered roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6e7cd9cebf073e63b840a5bc355f09318bd6c705"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-11 15:05:58.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\nThis patch also adds some fancy molecule testing.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1e8d4e79602284f6272bff429a993796e257e6b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1e8d4e79602284f6272bff429a993796e257e6b9"}]},"branch":"refs/heads/master"},"6f142de2c61681e4813506df3453932b8f877181":{"kind":"REWORK","_number":6,"created":"2020-02-19 08:03:47.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/6"}}},"commit":{"parents":[{"commit":"2a89e16cfeac23cce11c181f7097f8dead3ad92e","subject":"Add basic molecule tests for uncovered roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2a89e16cfeac23cce11c181f7097f8dead3ad92e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-19 08:02:19.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\nThis patch also adds some fancy molecule testing.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6f142de2c61681e4813506df3453932b8f877181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6f142de2c61681e4813506df3453932b8f877181"}]},"branch":"refs/heads/master"},"b026524f8ff7098d9ac811124dc1744e9c4d7de2":{"kind":"REWORK","_number":7,"created":"2020-02-19 12:33:56.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/29/705429/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/29/705429/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/29/705429/7"}}},"commit":{"parents":[{"commit":"2a89e16cfeac23cce11c181f7097f8dead3ad92e","subject":"Add basic molecule tests for uncovered roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2a89e16cfeac23cce11c181f7097f8dead3ad92e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-03 12:18:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-19 12:33:46.000000000","tz":60},"subject":"Updated/corrected service_status","message":"Updated/corrected service_status\n\nThis patch updates service_status check, importing what was done in\noooq-extras back in 2019[1].\n\nThis will allow to actually ensure only tripleo related services are\nchecked.\n\nThe legacy Docker support is still here in order to allow checking on\nolder releases.\n\nThis patch also adds some fancy molecule testing.\n\n[1] https://review.opendev.org/#/c/637729/\n\nChange-Id: I9016497305b4ec1a1e49b4b1c3cb23d219a42458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b026524f8ff7098d9ac811124dc1744e9c4d7de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b026524f8ff7098d9ac811124dc1744e9c4d7de2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
