)]}'
{"id":"openstack%2Ftripleo-validations~789698","triplet_id":"openstack%2Ftripleo-validations~master~Id047fe378cf512b985c9d7478a991f11b280102b","project":"openstack/tripleo-validations","branch":"master","topic":"automated_config_validation","hashtags":[],"change_id":"Id047fe378cf512b985c9d7478a991f11b280102b","subject":"Introducing the oslo_config_validator role","status":"MERGED","created":"2021-05-05 02:44:17.000000000","updated":"2021-06-30 07:35:37.000000000","submitted":"2021-06-18 16:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"789698-automated_config_validation","meta_rev_id":"a0e46507b0ff016f29b4f68cc0991f4ad6843409","_number":789698,"virtual_id_number":789698,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 16:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2021-06-16 15:01:09.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":2,"date":"2021-06-18 12:05:32.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":1,"date":"2021-06-18 12:05:32.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 02:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 04:07:34.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2021-05-19 04:07:34.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-05-19 04:07:34.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-05-31 14:22:48.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-06-18 12:05:32.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"d2e8e29a469f398325b8b373e3a3b4ac59cab938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-05 02:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99a7a97c30057061e1bbf88a28e1b0f2d0bd505a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-05 02:45:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b8dfb1572e1eca3ea38edfe15f1456b12c2b07af","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-05 02:46:09.000000000","message":"Topic set to automated_config_validation","accounts_in_message":[],"_revision_number":1},{"id":"b4d2c1dc6a721124a83baf95a4ff2ec9ec22dc8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 02:58:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/8e8a5680538c4dcab5313968ed644c96 : FAILURE in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/264d38b6920e45bc8703c80412bd677f : FAILURE in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb44a278fe446b48bc15367fc78367f : POST_FAILURE in 5m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2773198ea784d57934f8e46d0cf1678 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"037441414dd50e25ecdbcdc9eab27958de427f51","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-05 03:24:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"834f87edc99882f5ac755b138933ca9e6cb4e5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 03:50:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/e1f8fc67829c4429bb3c08464ce93c49 : RETRY_LIMIT in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dbb2870b8f14275a4749cffee7c6c53 : SUCCESS in 3m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/718bfcec06b44c50bcdc41cab0508e7c : RETRY_LIMIT in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb0ac0d884e47a8872b84df127283c4 : FAILURE in 4m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860ced6f3a6e4663ba45c87c5e154171 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0520964efa68252ae13fb84687571aaa7d29efe9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-10 19:21:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93af58e9c476b282c5c60db0b7713aba4cedb84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:37:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/ff01be4b7b074b28acd0041821895bc9 : RETRY_LIMIT in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae17b6415d794a7a97c28c60625fe4f6 : SUCCESS in 3m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52db572f26f4447ea069d8045c67898d : RETRY_LIMIT in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75d73a702614ebcae3de653c9b13b34 : SUCCESS in 6m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/991188559e3e42f684b93bca7e90dea9 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":3},{"id":"cd6a75460dadc6434700a1d404002171c6bdcd54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-11 00:37:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3a6c6382470f398c5230db621c091ddabff50578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 00:50:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/113fae7e43e748eb94ad412568d7dc4f : RETRY_LIMIT in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ebf73aa7a243cf8c326d47668d3580 : SUCCESS in 3m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9479721d0bc7487088a3baef857070be : RETRY_LIMIT in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c829b23e606447c9af0008e0cc8889d2 : SUCCESS in 7m 02s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b187055a39a64c529c8f87a065b9e95d : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"8ec91bc75aef23b5560e3942a04a7082ea67267c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-11 05:22:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4dea9e57fc2054dc96a5fbee8b5074082af11fb4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-11 05:28:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"41122152818259e8859a92363c51f66673996b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 05:42:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/580a119028b34031a584739a52495899 : FAILURE in 7m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8f34656796564e069552f939876bc387 : FAILURE in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94c5f08535f426ea0c249c9d089aefa : SUCCESS in 5m 01s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01b2ba3b9aa445d6a2ae4651d8890472 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":6},{"id":"0f440cad37381316c97f62674162d607ce29962e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-13 11:24:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cc19008dda98e4b7d0a98b628ff928550771f14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 11:41:20.000000000","message":"Patch Set 7: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/bfc157f970634660bbf1410940d6c701 : SUCCESS in 9m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4ff8ae2a2db43fbbd4fe3d3178e7d35 : FAILURE in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd324e107a347c788706a0d21f6c1da : SUCCESS in 6m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f486cb178d44b3fa830d4aafad240a0 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":7},{"id":"06fe83686696d42a983f01b17cbd2c7a255af13a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-13 12:18:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0173c9878bc0cd1665bd4754af16f0c755d6ac64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:02:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/d993d4af46a94f6c8b456b0ada92bb5c : SUCCESS in 7m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/796a9511ba464f4d9957367e8928534e : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae929dd425e143548a74668aae1bfc03 : SUCCESS in 6m 13s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7f2268b0c26b44f7a9059362faca8521 : SUCCESS in 2h 30m 39s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/93407602fcf4458dbcb0d4b37d06ef74 : SUCCESS in 1h 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03e763450aec48b7b129d80c1ca61f1a : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":8},{"id":"69783ba62ac6d6615d835ebcf1de3cc7f60395d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-15 06:03:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"23b5de66deddfb6381f847a112a424cbe47872ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-15 06:58:14.000000000","message":"Patch Set 9: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/d4c7a76f3dde4079b2acc923ea10214a : RETRY_LIMIT in 2m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6c929e83beb4b048bbc29437364a6bf : FAILURE in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ed7955f0224785badf4d1d64704641 : SUCCESS in 7m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be797f7baf904d2dab41681d970169a2 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":9},{"id":"bb3656283ce0a84915c010613ada5467daf9a89e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-16 03:59:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b306da320d1a96dfa9a56654c42931dbb1f71cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-16 04:17:57.000000000","message":"Patch Set 10: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/16d168678575429da100e984589e0063 : FAILURE in 7m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef55d2ede4bb4f6bb6333c1ff496a1e3 : FAILURE in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5f4920063c4db9ace2a8c94ecd9b4b : SUCCESS in 4m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850333dddd254a7a84ff6666dce380e0 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":10},{"id":"cea1b00d0043e5972b39a942317e869a404cb75e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-18 05:51:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b71b0c5e18beb7cf3eca44df01d17e4221324f5c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-18 05:51:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"c820526cfdca3eca4f0f3b05969683e35b33cae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 05:59:47.000000000","message":"Patch Set 11: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/312d5af9e3f7477793ec29a76fa4d3b1 : FAILURE in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61a91fe7196a4ceab8f9f26a2f031c9e : FAILURE in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52d5e28a6cd45c78443155c6592d93b : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adf38a76e9e240aa99962fb390e016d2 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":11},{"id":"4c91849a07bee954b51b419d01dff07b58a54723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-19 00:11:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c38eb4494dc6f24f55247b555985e201be767786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 03:28:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/7e5a16a07e704ac2a79b316b6fd4a5e8 : SUCCESS in 8m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dfc7967e8d4246b6b38eb9530ad3e10a : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53761a638e354f6289dcb8212a032b1b : SUCCESS in 5m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/db6ddd52dd5c45c8ae14fd214e03168e : SUCCESS in 2h 49m 48s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/1dfc9699345c48eaac53a3ef69153303 : SUCCESS in 1h 23m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5537e13a91a340ddb20d3ad33a9ebe65 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":12},{"id":"93cc6d8492a92f59585d4f5132e5b9953f40e9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-19 04:01:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"929144ca78c9887b7fcf6e11fd43625e2d7fbcfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-19 04:32:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f1eda3b38e48f80a2b32ef0d227b8cef92658ff1","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-19 04:35:30.000000000","message":"Patch Set 14:\n\nDavid, the PS#14 is just addressing the fact that we are now using a global molecule configuration at the project level since https://review.opendev.org/q/66193a69b79a51d436f132d30b0d530e0f1139ac","accounts_in_message":[],"_revision_number":14},{"id":"7c9f879dba8a7d84bdc3b0874b89d5077a7ea005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 05:04:59.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/fd09484b0f4e482b95e63163354023cf : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95b3e11fe3c8438a9f172a135a06925b : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93778653e884290b99f12ec9fe7e1bb : FAILURE in 6m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d451161eea8e4f28ae5a13affa98035d : FAILURE in 21m 39s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d8771b904744cdb630093ef1421395 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":14},{"id":"57af3abb6bc00deda95b155db873a7752586d56f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-19 18:52:26.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"44e7b6ee6b00b82a26539977e58ab679bb0de57c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:52:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/9351a20a415041958b58ba00a197c7eb : FAILURE in 7m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b2ba432338d442d8a098854ade32f7d5 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8276699613464197f2b97409a25f03 : SUCCESS in 5m 23s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d954dfc256854bc99f82bc1b1a24433a : SUCCESS in 2h 46m 43s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/0911752d2b36439bbb81e7da8964713b : SUCCESS in 1h 26m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df5a0600cab41fea8673b3417ec5faf : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":15},{"id":"c284aab10fea506ebf8091db4ff9a749b0229cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-20 04:24:19.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"4b5d12c21fc6cce4bdb1163dc322a89df7402b1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-20 04:33:28.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"4bd9bed8682252c4c4037c39d63353886a345642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:17:51.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/b17b37281f6e4f7598c1ce0e53031857 : SUCCESS in 9m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8261690d321746d79b57d6c053226cf2 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad3605fb15c4943b1e71c30a683d8fd : SUCCESS in 5m 31s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/26ecbd667cf1427e9c42b5deacf5add0 : SUCCESS in 2h 33m 22s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/6c84e0446dc34fc4a691ec50d05f61d4 : SUCCESS in 1h 28m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b73fcfe7bd42f4afb2891ac3cd8771 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":17},{"id":"2f98209420ecf17ee07349f9beebfb99d284f4d2","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-31 15:01:39.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)\n\nIt\u0027s better to err on the side of safety.","accounts_in_message":[],"_revision_number":17},{"id":"4fc2518049d26c59fe6e07888a94aa7b815158b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-31 15:15:36.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f16cb3d85d11c3ce64e1274df42c7201df90b91a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-31 15:20:27.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"336a2f552c807c6f80a08ca0f3f49ef525ff2012","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-31 15:21:38.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f1ea20842af6196d6e40f2130799dfc1f2d0d41f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-31 15:23:01.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7c2eb9b6e77d751e6aa1881ae83ef221dfa26cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 18:45:39.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/c4e70af46c7f478caff508b0feb49e0c : SUCCESS in 10m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9d06d4a390b8425aab2e7f1730607d07 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/779d395d89064d6b950ca88fae0be6d0 : SUCCESS in 6m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/057b784644fd434d8137adfc2965025f : SUCCESS in 3h 10m 49s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/0442238e2bea418f91ee8d547ac3a18a : SUCCESS in 1h 47m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cff3065f9494173b8065c5ad12b68eb : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d343404735684c328dc381b7c1fc7ecd : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":20},{"id":"25a1158f0063dfe85f2bce3b9cb511bdc5ef564f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-31 19:51:42.000000000","message":"Patch Set 20: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":20},{"id":"dc392cc76fd81023862e493cc0c66923fef2f18a","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-05-31 22:02:33.000000000","message":"Patch Set 20:\n\n(8 comments)","accounts_in_message":[],"_revision_number":20},{"id":"1fc7f3577be89fdf20143e76d00b05ae0361b44b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-01 05:44:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4b921fc24106366e5187a83df11e9b4fc5829e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 07:09:35.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/e221fa06ec5b42d1a862f227a9113d7c : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97d2507c820745eaab318fa66f6cc02b : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6408ad17764813b4dbd310204870b2 : SUCCESS in 6m 54s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4109f2bc04244514aba9193c8f15f1a6 : FAILURE in 1h 14m 08s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b717fd60147a41b8b7911c895098d27a : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcd0d0aac0cf4253b9e61761a4550fe3 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":21},{"id":"3ec3d4c4626c9dcb30a0d7a9b61886267a9efbbd","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-01 12:27:30.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"565e3579ec1f3bedc34e72143eeba603f5252508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:49:48.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/041d9c9662d2446dbe9491220cafd1f6 : SUCCESS in 10m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9227d710a254654b1ff877313d416c9 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ed19f32de34fb0b8a5f367da76bf84 : SUCCESS in 6m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5de3aa31f9ca4744ae52ab0ccccb9142 : SUCCESS in 4h 03m 58s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/a84ebe306f5940f086a804b52403625d : SUCCESS in 1h 19m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e91af456888a41d8beee2f8dedbef085 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abb2bdf89176487c8a0a623f6163afb6 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":21},{"id":"0e772299db0f55597ad3e9019687c1f2e1905c42","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-02 20:25:10.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"811c7034535a564aad8c0024d878bbf50e88e40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 04:24:30.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"57fec96a8aba06bc3e6ec00c4c1668b01a6bf816","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 04:24:46.000000000","message":"Patch Set 22: Workflow-1","accounts_in_message":[],"_revision_number":22},{"id":"5c4a034f83ad6e8d51ac86c37b553be92ea98dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 04:40:27.000000000","message":"Patch Set 22: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/1388576d523e4324a0d946905939323b : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a68c65bc090493cb7d7152bfc43f3da : FAILURE in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3687ddcf990c404396c728bef1589d91 : SUCCESS in 10m 03s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf073c8d17145aead649965da8d33db : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ad62b7574cd46d880f5677ee068ac84 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":22},{"id":"81e2bf40201cb1a095b4532c74c02809eacd1dd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 16:52:30.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"9a0931472a11cbaac2ed94df2c10bf47db761bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 16:57:21.000000000","message":"Patch Set 24: Commit message was updated.","accounts_in_message":[],"_revision_number":24},{"id":"9cf3f747cd874c34eb0d6e871bc9bf58ed581dc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 17:09:08.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"4c796fa09eb3324f0a29ec9d2eebb40b2eb90efb","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-06-04 17:29:39.000000000","message":"Patch Set 25:\n\n(4 comments)\n\nheading in the right direction; I mostly have nits that I picked\n\nPlease address these, and get a clean CI pass through, then I\u0027ll give a more thorough review.\n\nLooking good. :-)","accounts_in_message":[],"_revision_number":25},{"id":"3676b9144fea94b3378188bfa21ebb69662e08fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:32:52.000000000","message":"Patch Set 25: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/a6af5db14d7b46b5a85d767855def36a : SUCCESS in 10m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9eca8c18c53c46edbf3de63faa84b6fc : FAILURE in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c427e2994f344f5a8958bd3bc999f6b4 : SUCCESS in 6m 06s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa85f5e1fd64db29ca555e0dd0f8161 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1697cb097caf4d639f84a126dd8018a8 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":25},{"id":"fc768212a9e7fdd77e949bfa42c558ddcfe75108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 19:12:02.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"661fc19af7e9c1037768ea6c947edc07b76fe056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:28:30.000000000","message":"Patch Set 26: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/25076928b6104784801339c9737f9cae : SUCCESS in 10m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b517fd6187034971bf21774a90c5dcd6 : FAILURE in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0c07ecae084d5c9712608a051526d4 : SUCCESS in 6m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a84fd269774a96a468d3d34992220e : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6167ff94fd948e6b07d228394f644cc : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":26},{"id":"072f646ec7cc0ec08142c4ae6a10386686a350a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 19:46:59.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"64810d51d341a32427cc5ae321a4ac0c3be75064","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 19:48:03.000000000","message":"Patch Set 27:\n\n(4 comments)","accounts_in_message":[],"_revision_number":27},{"id":"905c27b596ab94d1de2185c1b6e24c7cc231955c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:47:01.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/eee063a4683e4bd192dad6ba4836f57a : SUCCESS in 11m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1e916fa1f6b403b9d1767c1ea1e2ec4 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9348f6a1d5a9431b84268d4079cda163 : SUCCESS in 5m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b7fc6ed52108401cbac6cb59cb6e5afd : SUCCESS in 2h 42m 56s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/f6e83d79110e4a70b13f9b1494099635 : SUCCESS in 1h 20m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb56065416b4bec8d2d758e35be3057 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f62bc21437724a4ab332a258e8668dc5 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":27},{"id":"3131879a4154db302b32f13a0580567df61067d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-07 13:57:24.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"8ea9e9385484b467e06cd806a9faca49dec03b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:58:48.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/a713c55a219443e39a051cf0f3904770 : SUCCESS in 10m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00fe0d0f8b344d3c9fb75db4f6746941 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd908fdb289848b5b4eeca0ebd43b579 : SUCCESS in 6m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9f78477b256a4b279cd680a95cce2acd : SUCCESS in 2h 18m 51s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/0996644177d746db93a94a4768ff7bca : SUCCESS in 1h 10m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86ab11984530494abd15da06929fc902 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a766ce9b8494f389f3745e57c292d0e : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":28},{"id":"b2744c7866a3d39da6f73ef4ea88f667079c5478","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-09 13:17:42.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"63dbba40bdbb4ec070b63cd0085387d39d234262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-11 14:59:17.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"43909b9ee906bf0e6a2cdcdefb9933abd1a24c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 21:21:26.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-validations-centos-8-molecule-oslo_config_validator https://zuul.opendev.org/t/openstack/build/99d2f431bbda45f99ec678fbe2c67b09 : SUCCESS in 12m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cfcff3a98e8b4084817e314ef75bbc53 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfda44b10b8a44b8a3f1fc2ae73cd455 : SUCCESS in 5m 40s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f698906673854d47b4322fa6b6a2e54f : SUCCESS in 2h 40m 31s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/9e54eacf666b49b3a355f2b39433e46d : SUCCESS in 1h 20m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c8630a6cad949459fc87912d63e6364 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d520857d658044618364cf2c04e3dd45 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":29},{"id":"2baba1109454d9eda335c5034a4402da0c5c5a15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-15 13:21:54.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f758eb76c540df9cf2a92392e69d5ed7df2cb37b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-15 13:28:38.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"a72c8e3c8af4f47aeaade1f36a76d528bb7aacbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:45:56.000000000","message":"Patch Set 31: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97effb1b76fc4c488f66de0ef52fc4fa : FAILURE in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e346ff82a444c7a70a29ae6fee3ed7 : SUCCESS in 7m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe60af0a01446e1832259b6a6506622 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/905517015752468ea54581e0877d1917 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":31},{"id":"b219b8541b28af6cc1d0e99fe095047516e25e58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-15 14:08:36.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"01586515db9720ebabfcc0e05053de1627a7b29d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-15 14:16:32.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"c15325820ad49a33e0361b9181872f307f06479b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:24:22.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a1c7b0485224740b8ddb23678df9edc : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b77f29a443743e381d9a66710b82351 : FAILURE in 7m 06s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/db56becc67bc4d5fa7054fe183e6ed1e : FAILURE in 40m 36s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d92ad10b3164798889b72e4395f6215 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a68973f302c44e3bb31408f8024df84 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":33},{"id":"98d2dfa6a93c680ca997265b581de4693b25d79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-15 17:17:59.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"c6b13c2aa4e26e418a9390c9fbf5e30dc2aeba2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:44:52.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/215c6bc44b7443199d422ae8d54fd303 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900bf2e17c7a4e2ea9fae29b5b7a559f : SUCCESS in 6m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ab9e0c9cca7a45fdb9757ac86db2b5ed : SUCCESS in 3h 09m 06s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/c5a22f544de049c1a358b9128365b6ef : FAILURE in 1h 18m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60cd2e4194c443e2b91be6f1e63df18d : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcca48d4df1641ccb8886b199257406a : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":34},{"id":"98a73557ba92ec739fdf10792d4aba6de9fd1081","tag":"compute.df.review","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-16 03:04:10.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"b487997f9f651fd957bf5cc803a0573c95ec1967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 05:49:45.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc96b4cce4404f72ad9180011333c192 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa976f1684c44bd8160cedd258821de : SUCCESS in 5m 50s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7fb41af2875548b7bf7e17673c657dbf : SUCCESS in 2h 35m 06s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/94584dcb519d4d6280c622b895bd7c35 : SUCCESS in 1h 19m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96926a3f48eb40a3952da44a25f7fa12 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e621ece21ef48bab1d631bea2c906a1 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":34},{"id":"54080ba21c0310ebc4f3e9af1caa0f7bab4b8172","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-16 15:01:09.000000000","message":"Patch Set 34: Code-Review+2","accounts_in_message":[],"_revision_number":34},{"id":"e81ee494d9ce32fdc6acb12db1779a1142b88a26","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-18 12:05:32.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1\n\nI was afraid by the podman_container.py and podman_container_info.py - but those are only mocked versions, so no real harm.","accounts_in_message":[],"_revision_number":34},{"id":"f70cf2bb846e3b77cf7198c38c7c8cf4f02b5165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:06:27.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"11c55c4dee66ede5e3b272bb8f0d74b2b7f21776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 16:02:51.000000000","message":"Patch Set 34: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4a83a381f5d9413ea2cb01a0d6901c6c : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/154b8124594948eab9774a97cfa027dc : SUCCESS in 6m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/adcdd48e6c3b441f8b1f4a95638887fe : SUCCESS in 3h 50m 43s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/81d6dbfd520d4499ada8f7c6c5237929 : SUCCESS in 1h 39m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1775e182dfef4bd6b0b3afdb68e18d04 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6d40ae2155a4eaeb8e14445a933c887 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":34},{"id":"da7c4cbf3fe203f2ed8d99297fa2a6b549fe5e6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 16:02:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":34},{"id":"bf4cc9ea3673c81f2a4462ea35f4780133d4bc13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 16:05:50.000000000","message":"Patch Set 34:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55016edaec66446b8a69337afba811bf : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":34},{"id":"d33a87763440e741dc13f8f396e4721cf7826e0b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-30 07:17:20.000000000","message":"Patch Set 34:\n\nJust noticed that the molecule job for that new role is defined in the zuul.d/molecule.yaml file but not executed in the CI!\n\nThis needs to be fixed","accounts_in_message":[],"_revision_number":34},{"id":"a0e46507b0ff016f29b4f68cc0991f4ad6843409","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-30 07:35:37.000000000","message":"Patch Set 34:\n\nWill be fix with https://review.opendev.org/c/openstack/tripleo-validations/+/797884","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"4aaca32f19af6426a6f1eb6ea896e8d10458928d","revisions":{"c360e7dbbd91086f410837ec361a81158ed9c179":{"kind":"REWORK","_number":1,"created":"2021-05-05 02:44:17.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/1"}}},"commit":{"parents":[{"commit":"ac70f2cef0c2ea99f8bb3bf99656b210dc23e5c9","subject":"Merge \"Exposing default variables in playbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ac70f2cef0c2ea99f8bb3bf99656b210dc23e5c9"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:43:54.000000000","tz":-240},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c360e7dbbd91086f410837ec361a81158ed9c179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c360e7dbbd91086f410837ec361a81158ed9c179"}]},"branch":"refs/heads/master"},"fa51dbb285a6f4c2ea94e4fd9eb5fd3ee3642857":{"kind":"REWORK","_number":2,"created":"2021-05-05 03:24:52.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/2"}}},"commit":{"parents":[{"commit":"ac70f2cef0c2ea99f8bb3bf99656b210dc23e5c9","subject":"Merge \"Exposing default variables in playbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ac70f2cef0c2ea99f8bb3bf99656b210dc23e5c9"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 03:24:45.000000000","tz":-240},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa51dbb285a6f4c2ea94e4fd9eb5fd3ee3642857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa51dbb285a6f4c2ea94e4fd9eb5fd3ee3642857"}]},"branch":"refs/heads/master"},"bc6b1383ea09f5dbca8b6df1d5ceef088d4245aa":{"kind":"REWORK","_number":3,"created":"2021-05-10 19:21:49.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/3"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-10 19:21:41.000000000","tz":-240},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bc6b1383ea09f5dbca8b6df1d5ceef088d4245aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bc6b1383ea09f5dbca8b6df1d5ceef088d4245aa"}]},"branch":"refs/heads/master"},"8b68497aaced9a90edc49a743da36c0092a646e1":{"kind":"REWORK","_number":4,"created":"2021-05-11 00:37:06.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/4"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-11 00:36:23.000000000","tz":0},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b68497aaced9a90edc49a743da36c0092a646e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b68497aaced9a90edc49a743da36c0092a646e1"}]},"branch":"refs/heads/master"},"46117a8e907cc3c44f4f5e68b295b463907f6ae5":{"kind":"REWORK","_number":5,"created":"2021-05-11 05:22:21.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/5"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-11 05:22:14.000000000","tz":-240},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/46117a8e907cc3c44f4f5e68b295b463907f6ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/46117a8e907cc3c44f4f5e68b295b463907f6ae5"}]},"branch":"refs/heads/master"},"5e7d40091f862ed8bc3757419ef5b8ca36ab59c0":{"kind":"REWORK","_number":6,"created":"2021-05-11 05:28:12.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/6"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-11 05:28:06.000000000","tz":-240},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5e7d40091f862ed8bc3757419ef5b8ca36ab59c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/5e7d40091f862ed8bc3757419ef5b8ca36ab59c0"}]},"branch":"refs/heads/master"},"327204c9d25b8785318088c0d5962c886fc34be3":{"kind":"REWORK","_number":7,"created":"2021-05-13 11:24:59.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/7"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-13 11:24:48.000000000","tz":0},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/327204c9d25b8785318088c0d5962c886fc34be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/327204c9d25b8785318088c0d5962c886fc34be3"}]},"branch":"refs/heads/master"},"370d36573e38278d3b1689f7327b66f43c02fee7":{"kind":"REWORK","_number":8,"created":"2021-05-13 12:18:57.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/8"}}},"commit":{"parents":[{"commit":"cee36599e147cdc3f2a52a1e11a14440f909fe4c","subject":"Merge \"Refactoring of ovs_dpdk_pmd_cpus_check tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cee36599e147cdc3f2a52a1e11a14440f909fe4c"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-05 02:42:00.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-13 12:18:48.000000000","tz":0},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/370d36573e38278d3b1689f7327b66f43c02fee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/370d36573e38278d3b1689f7327b66f43c02fee7"}]},"branch":"refs/heads/master"},"95865a4d04d559d168b6286958023c653f02200a":{"kind":"REWORK","_number":9,"created":"2021-05-15 06:03:48.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/9"}}},"commit":{"parents":[{"commit":"f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50","subject":"Merge \"Quick documentation content reorganization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/95865a4d04d559d168b6286958023c653f02200a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/95865a4d04d559d168b6286958023c653f02200a"}]},"branch":"refs/heads/master"},"6c22706bd2b5aeed19349cb2bcb070cb44297993":{"kind":"REWORK","_number":10,"created":"2021-05-16 03:59:57.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/10"}}},"commit":{"parents":[{"commit":"f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50","subject":"Merge \"Quick documentation content reorganization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-16 03:59:51.000000000","tz":0},"subject":"[WIP] Introducing the oslo_config_validator role","message":"[WIP] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6c22706bd2b5aeed19349cb2bcb070cb44297993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6c22706bd2b5aeed19349cb2bcb070cb44297993"}]},"branch":"refs/heads/master"},"acf9abd5fc1ff7e64583d1c486bb296f91b61a93":{"kind":"REWORK","_number":11,"created":"2021-05-18 05:51:46.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/11"}}},"commit":{"parents":[{"commit":"f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50","subject":"Merge \"Quick documentation content reorganization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-18 05:51:14.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/acf9abd5fc1ff7e64583d1c486bb296f91b61a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/acf9abd5fc1ff7e64583d1c486bb296f91b61a93"}]},"branch":"refs/heads/master"},"611c4243eb6bed9d0e65e454d2531eb324eab4f5":{"kind":"REWORK","_number":12,"created":"2021-05-19 00:11:50.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/12"}}},"commit":{"parents":[{"commit":"f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50","subject":"Merge \"Quick documentation content reorganization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-19 00:11:45.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/611c4243eb6bed9d0e65e454d2531eb324eab4f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/611c4243eb6bed9d0e65e454d2531eb324eab4f5"}]},"branch":"refs/heads/master"},"78e6f54a26e57f45e425159209b636dd896c4a71":{"kind":"REWORK","_number":13,"created":"2021-05-19 04:01:00.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/13"}}},"commit":{"parents":[{"commit":"f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50","subject":"Merge \"Quick documentation content reorganization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f8c5342b7fd04a6a90d67b10ff52010ef0c8aa50"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-19 04:00:53.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/78e6f54a26e57f45e425159209b636dd896c4a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/78e6f54a26e57f45e425159209b636dd896c4a71"}]},"branch":"refs/heads/master"},"4bf99ddb8a7eba8afd912d976fd7c90536d30060":{"kind":"REWORK","_number":14,"created":"2021-05-19 04:32:16.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/98/789698/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/14"}}},"commit":{"parents":[{"commit":"e7b320a7f785e485a41d5bb5e7279cac0fc0fb81","subject":"Merge \"Bump pre-commit repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e7b320a7f785e485a41d5bb5e7279cac0fc0fb81"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-05-19 04:32:01.000000000","tz":120},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4bf99ddb8a7eba8afd912d976fd7c90536d30060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4bf99ddb8a7eba8afd912d976fd7c90536d30060"}]},"branch":"refs/heads/master"},"c7b8abc4baa0c744203ad2ab81f13ce7d5fe1ce4":{"kind":"REWORK","_number":15,"created":"2021-05-19 18:52:26.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/15"}}},"commit":{"parents":[{"commit":"e7b320a7f785e485a41d5bb5e7279cac0fc0fb81","subject":"Merge \"Bump pre-commit repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e7b320a7f785e485a41d5bb5e7279cac0fc0fb81"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-19 18:52:20.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c7b8abc4baa0c744203ad2ab81f13ce7d5fe1ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c7b8abc4baa0c744203ad2ab81f13ce7d5fe1ce4"}]},"branch":"refs/heads/master"},"f45d15d8f487bec24b9cdda5dfa907dd956c8a03":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-05-20 04:24:19.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/98/789698/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/16"}}},"commit":{"parents":[{"commit":"a9a0ab6433bdbba43894e64e08cfcd347aa13207","subject":"Merge \"Remove workaround for tox-ansible and global molecule config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a0ab6433bdbba43894e64e08cfcd347aa13207"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2021-05-20 04:24:19.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f45d15d8f487bec24b9cdda5dfa907dd956c8a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f45d15d8f487bec24b9cdda5dfa907dd956c8a03"}]},"branch":"refs/heads/master","description":"Rebase"},"6884f12c96732cf86181e79eb238429bbda49dfc":{"kind":"REWORK","_number":17,"created":"2021-05-20 04:33:28.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/98/789698/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/17"}}},"commit":{"parents":[{"commit":"a9a0ab6433bdbba43894e64e08cfcd347aa13207","subject":"Merge \"Remove workaround for tox-ansible and global molecule config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a0ab6433bdbba43894e64e08cfcd347aa13207"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2021-05-20 04:31:45.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6884f12c96732cf86181e79eb238429bbda49dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6884f12c96732cf86181e79eb238429bbda49dfc"}]},"branch":"refs/heads/master"},"0752e30db59794984c102ee574529e17cd579d4a":{"kind":"REWORK","_number":18,"created":"2021-05-31 15:15:36.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/18"}}},"commit":{"parents":[{"commit":"a9a0ab6433bdbba43894e64e08cfcd347aa13207","subject":"Merge \"Remove workaround for tox-ansible and global molecule config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a0ab6433bdbba43894e64e08cfcd347aa13207"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-31 15:15:31.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0752e30db59794984c102ee574529e17cd579d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0752e30db59794984c102ee574529e17cd579d4a"}]},"branch":"refs/heads/master"},"795dac1d77a2c222eeddbc2b191b26e77d5360df":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-05-31 15:20:27.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/19"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-31 15:20:27.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/795dac1d77a2c222eeddbc2b191b26e77d5360df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/795dac1d77a2c222eeddbc2b191b26e77d5360df"}]},"branch":"refs/heads/master","description":"Rebase"},"a942855f38872c6446f48fc881518669c95d3288":{"kind":"REWORK","_number":20,"created":"2021-05-31 15:23:01.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/20"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-31 15:22:56.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a942855f38872c6446f48fc881518669c95d3288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a942855f38872c6446f48fc881518669c95d3288"}]},"branch":"refs/heads/master"},"fa456e1031b395cadd1ecc3bb97dd1ad219b65aa":{"kind":"REWORK","_number":21,"created":"2021-06-01 05:44:01.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/21"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-15 06:03:33.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-01 05:43:50.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa456e1031b395cadd1ecc3bb97dd1ad219b65aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa456e1031b395cadd1ecc3bb97dd1ad219b65aa"}]},"branch":"refs/heads/master"},"a1514c1820730b6bc0e2d9f349e4ca4d2342bd2e":{"kind":"REWORK","_number":22,"created":"2021-06-04 04:24:30.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/22"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"subject":"[DNM] Introducing the oslo_config_validator role","message":"[DNM] Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing it will also validate the values of some specific settings.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a1514c1820730b6bc0e2d9f349e4ca4d2342bd2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a1514c1820730b6bc0e2d9f349e4ca4d2342bd2e"}]},"branch":"refs/heads/master"},"181ea9d9f2330373cc485aeab80b9c7670cc1db5":{"kind":"REWORK","_number":23,"created":"2021-06-04 16:52:30.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/23"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 16:51:06.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/181ea9d9f2330373cc485aeab80b9c7670cc1db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/181ea9d9f2330373cc485aeab80b9c7670cc1db5"}]},"branch":"refs/heads/master"},"de00bc3b5664fa0c25f6706fd855aa61ab6a208b":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2021-06-04 16:57:21.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/24"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 16:57:21.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nThis should be easily adaptable in train to validate settings during\nhybrid mode with dockerized overcloud.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/de00bc3b5664fa0c25f6706fd855aa61ab6a208b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/de00bc3b5664fa0c25f6706fd855aa61ab6a208b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c6c2d2a5fc2b4f0830aec06c440d8c8ed4019900":{"kind":"REWORK","_number":25,"created":"2021-06-04 17:09:08.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/25"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 17:08:49.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c6c2d2a5fc2b4f0830aec06c440d8c8ed4019900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c6c2d2a5fc2b4f0830aec06c440d8c8ed4019900"}]},"branch":"refs/heads/master"},"fdfcec7a0cd9a8e93a65ec1d402791bdce6cecc6":{"kind":"REWORK","_number":26,"created":"2021-06-04 19:12:02.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/26"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 19:11:52.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fdfcec7a0cd9a8e93a65ec1d402791bdce6cecc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fdfcec7a0cd9a8e93a65ec1d402791bdce6cecc6"}]},"branch":"refs/heads/master"},"eff55e56f9c14fc1b418517ad6ba509937285041":{"kind":"REWORK","_number":27,"created":"2021-06-04 19:46:59.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/27"}}},"commit":{"parents":[{"commit":"a9a316654a72f386aa3afb58298c0cab34360fab","subject":"Merge \"Removing python-tripleoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a9a316654a72f386aa3afb58298c0cab34360fab"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 19:46:53.000000000","tz":0},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/eff55e56f9c14fc1b418517ad6ba509937285041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/eff55e56f9c14fc1b418517ad6ba509937285041"}]},"branch":"refs/heads/master"},"371e27bb098e39304c133531e6820a1062f63d70":{"kind":"REWORK","_number":28,"created":"2021-06-07 13:57:24.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/28"}}},"commit":{"parents":[{"commit":"26159d9d78060c1b44ba9e362123f86e4021ac2b","subject":"Merge \"Introduce new Ansible validation_init role.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/26159d9d78060c1b44ba9e362123f86e4021ac2b"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-07 13:57:13.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/371e27bb098e39304c133531e6820a1062f63d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/371e27bb098e39304c133531e6820a1062f63d70"}]},"branch":"refs/heads/master"},"9eef4278d83739e30b06c759a7dc520aa4c62b35":{"kind":"REWORK","_number":29,"created":"2021-06-11 14:59:17.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/29"}}},"commit":{"parents":[{"commit":"0d7c3f49522e3bb4ce756e6c7892c722e6ec49a3","subject":"Merge \"Use OS_CLOUD environment variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0d7c3f49522e3bb4ce756e6c7892c722e6ec49a3"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-11 14:58:53.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9eef4278d83739e30b06c759a7dc520aa4c62b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9eef4278d83739e30b06c759a7dc520aa4c62b35"}]},"branch":"refs/heads/master"},"e0643b9ed0effb0c56fc22bc97b425d4a1999c0d":{"kind":"REWORK","_number":30,"created":"2021-06-15 13:21:54.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/30"}}},"commit":{"parents":[{"commit":"49e54f46855b35ae2e8498ff284ca63ffc156f84","subject":"Merge \"Allow comments in variable files to be rendered in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49e54f46855b35ae2e8498ff284ca63ffc156f84"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-15 13:21:40.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0643b9ed0effb0c56fc22bc97b425d4a1999c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0643b9ed0effb0c56fc22bc97b425d4a1999c0d"}]},"branch":"refs/heads/master"},"10c812cdd534d2d65c8ab7ecf28399610fca78d1":{"kind":"REWORK","_number":31,"created":"2021-06-15 13:28:38.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/31"}}},"commit":{"parents":[{"commit":"49e54f46855b35ae2e8498ff284ca63ffc156f84","subject":"Merge \"Allow comments in variable files to be rendered in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49e54f46855b35ae2e8498ff284ca63ffc156f84"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-15 13:28:32.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/10c812cdd534d2d65c8ab7ecf28399610fca78d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/10c812cdd534d2d65c8ab7ecf28399610fca78d1"}]},"branch":"refs/heads/master"},"179ae015a8e96acb0a1f27012d0d505a8685587e":{"kind":"REWORK","_number":32,"created":"2021-06-15 14:08:36.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/32"}}},"commit":{"parents":[{"commit":"49e54f46855b35ae2e8498ff284ca63ffc156f84","subject":"Merge \"Allow comments in variable files to be rendered in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49e54f46855b35ae2e8498ff284ca63ffc156f84"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-15 14:08:31.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/179ae015a8e96acb0a1f27012d0d505a8685587e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/179ae015a8e96acb0a1f27012d0d505a8685587e"}]},"branch":"refs/heads/master"},"b48569e3324954dc4b7b095fe61f5362dcca0807":{"kind":"REWORK","_number":33,"created":"2021-06-15 14:16:32.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/33"}}},"commit":{"parents":[{"commit":"49e54f46855b35ae2e8498ff284ca63ffc156f84","subject":"Merge \"Allow comments in variable files to be rendered in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49e54f46855b35ae2e8498ff284ca63ffc156f84"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-15 14:16:20.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b48569e3324954dc4b7b095fe61f5362dcca0807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b48569e3324954dc4b7b095fe61f5362dcca0807"}]},"branch":"refs/heads/master"},"4aaca32f19af6426a6f1eb6ea896e8d10458928d":{"kind":"REWORK","_number":34,"created":"2021-06-15 17:17:59.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/98/789698/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/98/789698/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/98/789698/34"}}},"commit":{"parents":[{"commit":"49e54f46855b35ae2e8498ff284ca63ffc156f84","subject":"Merge \"Allow comments in variable files to be rendered in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/49e54f46855b35ae2e8498ff284ca63ffc156f84"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 04:24:24.000000000","tz":0},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-15 17:17:54.000000000","tz":-240},"subject":"Introducing the oslo_config_validator role","message":"Introducing the oslo_config_validator role\n\nThis role is intended to leverage the `oslo-config-validator` on each one\nof the configuration files found on a deployment. The goal is to quickly\ncatch erroneous configurations.\n\nWhen called manually, it will also be possible to generate a report\nreturning all the differences between the current configuration and the\ndefault configuration\n\nOne last thing, it will also validate the values of some specific\nsettings, based on the invalid_settings dict in the configuration.\nBecause of that, we moved some checks in nova_event_callback to the\noslo_config_validator role.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.config/+/790883\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1940815\nResolves: rhbz#1940815\nChange-Id: Id047fe378cf512b985c9d7478a991f11b280102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4aaca32f19af6426a6f1eb6ea896e8d10458928d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4aaca32f19af6426a6f1eb6ea896e8d10458928d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
