)]}'
{"id":"openstack%2Ftripleo-validations~795749","triplet_id":"openstack%2Ftripleo-validations~stable%2Fwallaby~I1fb32100139736b88a078940f930122f2694955c","project":"openstack/tripleo-validations","branch":"stable/wallaby","hashtags":["DEV_WORKFLOW_WALLABY"],"change_id":"I1fb32100139736b88a078940f930122f2694955c","subject":"Add Validation Development Workflow Documentation","status":"MERGED","created":"2021-06-10 09:08:41.000000000","updated":"2021-06-18 04:45:16.000000000","submitted":"2021-06-18 04:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795749","meta_rev_id":"10c4f450dcaa35d23801c37044a07fe29715817d","_number":795749,"virtual_id_number":795749,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 04:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"all":[{"value":2,"date":"2021-06-10 12:43:54.000000000","_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2021-06-17 19:39:30.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":1,"date":"2021-06-12 14:50:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-10 09:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":1,"date":"2021-06-17 19:39:30.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 09:09:52.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-10 09:09:52.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-10 09:09:52.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-10 09:09:52.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-10 09:09:52.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-06-10 09:29:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-12 14:50:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REVIEWER"}],"messages":[{"id":"2bd92589bdc85fb3b5a8d528bb7e38ce7f018f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-10 09:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96cf9345af66671f9acae4e360a8b8286b04b636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 09:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/249fd56058e54981b2fa735c053100be : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894ce13af2d34dc1a4dc1128fe32f390 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"28663930167dc4290fd1840a189dfea03c2ed27d","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-10 09:46:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4dabf7e0c59ab008a26fbe8241f93244858eb473","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-10 12:04:44.000000000","message":"Hashtag added: DEV_WORKFLOW_WALLABY","accounts_in_message":[],"_revision_number":1},{"id":"e170ada42fb4ccef0c74e66f38d020ad75d01155","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-06-10 12:43:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51c925a8078ce1664fe0d0f4a77dfef6b7d9f4c5","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-12 14:50:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3236cc75d13977e92fe2cd083e7997affc904ea5","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-06-17 19:39:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b67041a7fabf8f41ca0cb995844d6d5f7c4259ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 00:55:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67f4769eac2393d001221832d3b618219b057d14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 04:42:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/319218cdb8b347da8bf7bf0dea2cb6a3 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c36ec847ce64bd5afde8155f5f68a44 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2f9c0da41c5443d95f45789bc185f2ac9b3c668d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 04:42:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10c4f450dcaa35d23801c37044a07fe29715817d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 04:45:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc075c7077924a69835decb6a45b3a1e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99acc1cb493a7c6a7c70b776d9cc5a6aa5944256","revisions":{"99acc1cb493a7c6a7c70b776d9cc5a6aa5944256":{"kind":"REWORK","_number":1,"created":"2021-06-10 09:08:41.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/49/795749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/49/795749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/49/795749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/49/795749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/49/795749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/49/795749/1"}}},"commit":{"parents":[{"commit":"19bb19ea3c74feec3ceb6e6c31b14ca5855dae16","subject":"Introduce new Ansible validation_init role.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/19bb19ea3c74feec3ceb6e6c31b14ca5855dae16"}]}],"author":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-05-27 13:49:12.000000000","tz":120},"committer":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-10 09:08:13.000000000","tz":120},"subject":"Add Validation Development Workflow Documentation","message":"Add Validation Development Workflow Documentation\n\nThis patch adds documentation for:\n- How to develop a molecule test for a new role\n- How to run the molecule test scenarios via tox-ansible\n\nSigned-off-by: Gael Chamoulaud (Strider) \u003cgchamoul@redhat.com\u003e\nChange-Id: I1fb32100139736b88a078940f930122f2694955c\n(cherry picked from commit 443f8d328f63df4cc65c0ebf84c717772e77ed31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/99acc1cb493a7c6a7c70b776d9cc5a6aa5944256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/99acc1cb493a7c6a7c70b776d9cc5a6aa5944256"}]},"parents_data":[{"branch_name":"refs/heads/stable/wallaby","commit_id":"19bb19ea3c74feec3ceb6e6c31b14ca5855dae16","is_merged_in_target_branch":true,"change_id":"I219f4054e6b854306719b9c91a0738c526f07666","change_number":795098,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
