)]}'
{"id":"openstack%2Ftripleo-validations~796509","triplet_id":"openstack%2Ftripleo-validations~stable%2Fwallaby~Ia8b78c0e81d716718139aec17f959a1a9955025e","project":"openstack/tripleo-validations","branch":"stable/wallaby","hashtags":[],"change_id":"Ia8b78c0e81d716718139aec17f959a1a9955025e","subject":"Use and ownership info for the check_flavors module","status":"MERGED","created":"2021-06-15 15:00:41.000000000","updated":"2021-06-21 12:44:06.000000000","submitted":"2021-06-21 12:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796509","meta_rev_id":"629fac15b79fddfe90fbaa84a3271fdd5bc35044","_number":796509,"virtual_id_number":796509,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 12:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2021-06-21 09:19:52.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 08:35:57.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2021-06-21 09:19:52.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 15:03:36.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-15 15:03:36.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-15 15:03:36.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-15 15:03:36.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-15 15:03:36.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-06-15 15:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84c9c7cd79ce5de10a6f958614b3e2127540dbd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-15 15:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4b58d328659d679a8a83e2c6a7dfcd6f62b4cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:22:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ba5629dd77443169a0544890f669593 : FAILURE in 21m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78e7e3387820495e9035ac1b210bc74e : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9da850689d4e46fa94a944c9a44517b761db89e3","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-16 05:37:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0eaa3dd88208933e0766ccdf2f487de650e0fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:02:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/840cffb2c52c47bea2c9d2508d487bfe : SUCCESS in 8m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a643580f26d14a159c66e41226a7ae60 : FAILURE in 3h 11m 30s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5175189e117b47ef9fcb523cb0fffbf5 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7dfb6119acd7cebf5ee76291dcb90a07767005a0","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-17 08:35:57.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0a1c8e40a38c595e7c53b901e44b52d0cd2f107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:32:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dcde90cac7fb4ec9bd536fb599dfc104 : SUCCESS in 8m 48s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3df570f9a229447eb9ac2b78d8fb19ef : SUCCESS in 2h 45m 20s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/99833f91f338411d93d66ef3cf37fc71 : SUCCESS in 1h 26m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d1be24b4db4a098a6172c568af449d : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5793894b0f753048fbab892fa4fac7f980cfd0e4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-21 09:19:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67139545815de8aacb089b1a24fdd8125e931c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:20:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeb3b484865d6b7afe528aefe246d788f16cce5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:44:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e10c4808f6274a83bda44bdcd497bc54 : SUCCESS in 7m 50s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8c41082cd9904c9fa0dff2ced307e462 : SUCCESS in 3h 14m 36s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/cd2a58347da5445181990f290f106896 : SUCCESS in 1h 51m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fc9be478d9e41c5a82b55c904bc5c4d : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"629fac15b79fddfe90fbaa84a3271fdd5bc35044","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:44:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"149b72e5999b7191d61b4d2f84b8c462cde4bcd9","revisions":{"149b72e5999b7191d61b4d2f84b8c462cde4bcd9":{"kind":"REWORK","_number":1,"created":"2021-06-15 15:00:41.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/09/796509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/09/796509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/09/796509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/09/796509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/09/796509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/09/796509/1"}}},"commit":{"parents":[{"commit":"05c0a56278da75cb962a37528f011a7d8f768add","subject":"Merge \"Remove the post-update groups.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/05c0a56278da75cb962a37528f011a7d8f768add"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-02 09:52:53.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-15 15:00:34.000000000","tz":120},"subject":"Use and ownership info for the check_flavors module","message":"Use and ownership info for the check_flavors module\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia8b78c0e81d716718139aec17f959a1a9955025e\n(cherry picked from commit 81bd49d628c50398c9f001f2b20ae3060d2b7835)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/149b72e5999b7191d61b4d2f84b8c462cde4bcd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/149b72e5999b7191d61b4d2f84b8c462cde4bcd9"}]},"parents_data":[{"branch_name":"refs/heads/stable/wallaby","commit_id":"05c0a56278da75cb962a37528f011a7d8f768add","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
