)]}'
{"id":"openstack%2Ftripleo-validations~801093","triplet_id":"openstack%2Ftripleo-validations~master~I4ded939b007f979ec81c2389ac1912115c555c3c","project":"openstack/tripleo-validations","branch":"master","topic":"collections-abc","hashtags":[],"change_id":"I4ded939b007f979ec81c2389ac1912115c555c3c","subject":"Replace deprecated import of ABCs from collections","status":"MERGED","created":"2021-07-16 16:04:50.000000000","updated":"2021-07-20 11:18:35.000000000","submitted":"2021-07-20 11:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801093-collections-abc","meta_rev_id":"f0fccf33df98467db1c566b1b944d85c6e50e3f2","_number":801093,"virtual_id_number":801093,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 11:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-07-20 06:32:30.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-07-19 08:42:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":2,"date":"2021-07-20 06:57:48.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-19 14:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2021-07-20 06:57:48.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 18:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-07-19 06:40:40.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-07-19 08:42:42.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"}],"messages":[{"id":"20231faf6f1c44872dc57e4d1fc7c62f09d6f35b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 16:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78fa83d8308522e2e8846ba77c44d05bbbcffd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 18:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0358109b14044de9b6aeadbf899e529a : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988c9671dd4a4612a40b82e324d59091 : SUCCESS in 5m 54s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/bced9c25720e4305967d1c387db5ddc3 : SUCCESS in 2h 38m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf1a3a31409444b4a439a3002db42280 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b13400711302430db689838a61964212 : SUCCESS in 5m 31s\n- tripleo-validations-coverchange https://zuul.opendev.org/t/openstack/build/ecd2ff86e03746bbac9e21a6d2a6ff0c : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8bcafa85423924a46bbbc3c07f9720e79012c7d","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-19 06:40:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ce30f85eb2c3dd9a474f74209e88bbd23143d35","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-19 07:29:33.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm zuul is happy and grep said you didn\u0027t miss any collections.Mapping or collections.Iterable ;)","accounts_in_message":[],"_revision_number":1},{"id":"8c9ee8be1686f1eb648fccc4a599f7b0cfda266b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-19 07:32:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ab66d0471d22c89ae77b913767cc5267fa4c53a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 07:53:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e979a60ad7226bb6d25e6d5b4f33a02c55a37340","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 07:54:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdc6c2fd91abcea95a4b4c61fc8901f5629d07de","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2021-07-19 08:42:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3df57cb28013100b345d566a0fa5fd0538c6d56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 10:00:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4cbfc1da99f4d2aa2b59c6bdeb2a645 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4adb2e10a814e2faf7f5b3bf0b09c11 : SUCCESS in 5m 34s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/d99ad472acdd4adfb1235994221d76ab : SUCCESS in 1h 56m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4249e99862454a1eb7dd156aa2540a5d : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8beb2b0288c4de0969839f73f9d1a05 : SUCCESS in 5m 30s\n- tripleo-validations-coverchange https://zuul.opendev.org/t/openstack/build/9faaf8672287428395487545d5c726d8 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b588bcc37f4c3ca3b1bde43f3aa71d3ad7919333","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-19 14:09:22.000000000","message":"Patch Set 2: Code-Review+1\n\nNot many VF cores around today. So this may take a while to get in.","accounts_in_message":[],"_revision_number":2},{"id":"508d2b3956804f8ed2a6412ae8aade96392c55f9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-20 06:32:30.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks for update","accounts_in_message":[],"_revision_number":2},{"id":"68b5a8f911e1817945a3c5d2719bd57b5e5deed4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-07-20 06:57:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9884e2f1f1fb51a33a686989fd872f38c74a016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 06:58:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aacc6bc1c767645584e23b04d69db8452f74e829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 11:17:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00a8e63c32704452af5d9de067e5226b : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11bbbceb55cf4844b8e8a6d314cf293c : SUCCESS in 5m 56s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/13673590c21e48f0a854294840a8848e : SUCCESS in 2h 24m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5beee3c085bc4e928e66f03c455b6efe : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c6fedba0ad540cfa45f3a5e5b274213 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"526e423a019df87af8a1f908975bcba6cc5d2bcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 11:17:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f0fccf33df98467db1c566b1b944d85c6e50e3f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 11:18:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2869788e1ce4f07b34687060cacc807 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9196b98ff1c3dd261385e8280f8254b4db8a4989","revisions":{"1b7e9694931528c9eaedab1f622c982eab3de8cc":{"kind":"REWORK","_number":1,"created":"2021-07-16 16:04:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/801093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/93/801093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/1"}}},"commit":{"parents":[{"commit":"8e2f503244120013920df88f2dbdaff8784f79f0","subject":"Merge \"Removing import conditional for \u0027collections.abc\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8e2f503244120013920df88f2dbdaff8784f79f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 16:04:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 16:04:27.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nChange-Id: I4ded939b007f979ec81c2389ac1912115c555c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1b7e9694931528c9eaedab1f622c982eab3de8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1b7e9694931528c9eaedab1f622c982eab3de8cc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8e2f503244120013920df88f2dbdaff8784f79f0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9196b98ff1c3dd261385e8280f8254b4db8a4989":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-19 07:53:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/801093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/93/801093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/93/801093/2"}}},"commit":{"parents":[{"commit":"8e2f503244120013920df88f2dbdaff8784f79f0","subject":"Merge \"Removing import conditional for \u0027collections.abc\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8e2f503244120013920df88f2dbdaff8784f79f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 16:04:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 07:52:21.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3[1].\n\n[1] https://docs.python.org/3/library/collections.html\n---\nDeprecated since version 3.3, will be removed in version 3.10: Moved\nCollections Abstract Base Classes to the collections.abc module.\nFor backwards compatibility, they continue to be visible in this module\nthrough Python 3.9.\n---\n\nChange-Id: I4ded939b007f979ec81c2389ac1912115c555c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9196b98ff1c3dd261385e8280f8254b4db8a4989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9196b98ff1c3dd261385e8280f8254b4db8a4989"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8e2f503244120013920df88f2dbdaff8784f79f0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
