)]}'
{"id":"openstack%2Ftripleo-validations~809663","triplet_id":"openstack%2Ftripleo-validations~stable%2Ftrain~I4584bca6385b44e4b4399a1146d0983dc078644b","project":"openstack/tripleo-validations","branch":"stable/train","topic":"victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"I4584bca6385b44e4b4399a1146d0983dc078644b","subject":"Bufix: Fix typo","status":"ABANDONED","created":"2021-09-20 13:36:50.000000000","updated":"2021-10-06 13:03:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809662,"cherry_pick_of_patch_set":3,"meta_rev_id":"3bdd5ffac7791415848f214f5b4d9d5ff9a53627","_number":809663,"virtual_id_number":809663,"owner":{"_account_id":29296,"name":"Uemit Seren","email":"uemit.seren@gmail.com","username":"timeu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-20 16:09:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 16:09:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c28031b1e0aec5a610e655ab43e825fab1a5eeaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29296,"name":"Uemit Seren","email":"uemit.seren@gmail.com","username":"timeu"},"date":"2021-09-20 13:36:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"07dd2edf33f4bd7736e03a280e554f8f2f4f43e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 16:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad74f2938554d0581aa8abb32b4261e : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27bc60fcd78f4617b635f98f17e4a5a2 : SUCCESS in 7m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/03de4cf146e241d18d6cda614558d2f3 : SUCCESS in 11m 20s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/17f2e371454c44a58c1a8c02e676e52a : SUCCESS in 2h 15m 09s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/f4de2e579962496eacecf5f3d643b7cf : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3bdd5ffac7791415848f214f5b4d9d5ff9a53627","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-10-06 13:03:54.000000000","message":"Abandoned\n\nFixed by another patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3399f15a819950c9608ab85a740913c09f6b5189","revisions":{"3399f15a819950c9608ab85a740913c09f6b5189":{"kind":"REWORK","_number":1,"created":"2021-09-20 13:36:50.000000000","uploader":{"_account_id":29296,"name":"Uemit Seren","email":"uemit.seren@gmail.com","username":"timeu"},"ref":"refs/changes/63/809663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/63/809663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/63/809663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/63/809663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/63/809663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/63/809663/1"}}},"commit":{"parents":[{"commit":"af96faba5e289c9a47070a654787c17b31b3bb48","subject":"Merge \"remove using containers-prepare-parameter.yaml\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/af96faba5e289c9a47070a654787c17b31b3bb48"}]}],"author":{"name":"Uemit Seren","email":"uemit.seren@gmail.com","date":"2021-09-20 12:35:57.000000000","tz":120},"committer":{"name":"Uemit Seren","email":"uemit.seren@gmail.com","date":"2021-09-20 13:36:50.000000000","tz":0},"subject":"Bufix: Fix typo","message":"Bufix: Fix typo\n\nThis was causing the nova-event-callback validation\nwhich is run as part of the post-deployment validation to fail:\n\"The task includes an option with an undefined variable. The error was: \u0027vig_plugging_timeout_value\u0027 is undefined\"\n\nChange-Id: I4584bca6385b44e4b4399a1146d0983dc078644b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3399f15a819950c9608ab85a740913c09f6b5189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3399f15a819950c9608ab85a740913c09f6b5189"}]},"parents_data":[{"branch_name":"refs/heads/stable/train","commit_id":"af96faba5e289c9a47070a654787c17b31b3bb48","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
