)]}'
{"id":"openstack%2Ftripleo-validations~813764","triplet_id":"openstack%2Ftripleo-validations~stable%2Fvictoria~I6f86ce6b135be9841b5fcae1f940c829327a2db2","project":"openstack/tripleo-validations","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"17216":{"account":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"last_update":"2021-10-14 03:55:20.000000000","reason":"Change was submitted"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2021-10-14 03:55:20.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2021-10-14 03:55:20.000000000","reason":"Change was submitted"},"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2021-10-13 20:01:19.000000000","reason":"Gael Chamoulaud replied on the change"},"27419":{"account":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"last_update":"2021-10-14 03:55:20.000000000","reason":"Change was submitted"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2021-10-14 03:55:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f86ce6b135be9841b5fcae1f940c829327a2db2","subject":"partial-backport: oslo-config-validator: Removing merged patches","status":"MERGED","created":"2021-10-13 14:56:47.000000000","updated":"2021-10-14 03:55:20.000000000","submitted":"2021-10-14 03:55:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813764","cherry_pick_of_change":813340,"cherry_pick_of_patch_set":2,"meta_rev_id":"4d36d53dcb15fb910f26e21bc575ad302902ac76","_number":813764,"virtual_id_number":813764,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-14 03:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2021-10-13 20:01:19.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2021-10-13 20:01:19.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-10-13 14:56:47.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bfff0265fed53533ce580a006f56617aec47c70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-13 14:56:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* roles/oslo_config_validator/molecule/default/molecule.yml\n* roles/oslo_config_validator/resources/library/podman","accounts_in_message":[],"_revision_number":1},{"id":"06d9e4fb35ef1db14625ff43836322ed1dcb784c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:04:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9a76d08b4b4e68a8e19cccd5910207 : FAILURE in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8465f5aeb09c4d5f86fe7e34357372ee : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a13ee4654349e6860374a36e757fec : FAILURE in 4m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/695304abdd554406b75c31eb62c4a1b6 : FAILURE in 6m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"d8858ceff3bc52c6df7f92635252a7fd2848a8bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-13 15:08:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f885590993c40101451b5f2cd8c00e57a4b519e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-13 15:10:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"62d4a380ad935cda4044bf8e3b6b7ddc929e0d91","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-13 15:11:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"278b98e0894679d01d3061f9df34726224c76b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 18:11:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7644bd290cef427e9dd63f8682dfb27b : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/152c663185bc41fc8b843c621faa3576 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d76abbee8e5747cbafca15f99074a9d1 : SUCCESS in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/80229fef658742b7b1cef2142e89d301 : SUCCESS in 10m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/897ce354bcf4491c92c2fc030eb48bdd : SUCCESS in 2h 35m 47s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/5c254a00a0c14444992a7a69152d9a59 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":3},{"id":"5846687c6e9b1a08587fedb29939c1aa491be0af","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-10-13 20:01:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fc1e79c3e86654b0fa86fed690aa5ebc3b30286c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 20:02:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d2f57c8dd70feeef11a83625282b463c17c015f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 23:09:13.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f6fae890f4400beda953e84d01dc9bc0b5d7823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 03:55:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299b0bfd210f45b5918efd66a7dc4f63 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53c65dbd24044d9fb1e1350f44f99189 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9dda889c1134bfdb904500be262bfdc : SUCCESS in 7m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b7e2cbda54d4d27a91b4677c99e95af : SUCCESS in 7m 53s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/528e394b84b54eafafc1876fb324c81c : SUCCESS in 2h 53m 15s\n- tripleo-ci-centos-8-standalone-tv-validation https://zuul.opendev.org/t/openstack/build/e8dc5e73256c4662a51e39a04cd0ca33 : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4d36d53dcb15fb910f26e21bc575ad302902ac76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 03:55:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75b0fc7dced7e0c5811be5b4d9fd60e0835f58a5","revisions":{"03c3612a89fd0fdf45dd41f4fc16f2e68d11bb22":{"kind":"REWORK","_number":1,"created":"2021-10-13 14:56:47.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/64/813764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/64/813764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/1"}}},"commit":{"parents":[{"commit":"d925c56f8dd36b3b814d2b4cca9faac83a7db47f","subject":"Rename system_encoding playbooks.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d925c56f8dd36b3b814d2b4cca9faac83a7db47f"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-05 02:35:35.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-13 14:56:47.000000000","tz":0},"subject":"oslo-config-validator: Removing merged patches","message":"oslo-config-validator: Removing merged patches\n\nWhile developing this, we discovered a lot of inconsistencies across the\nopenstack projects. These inconsistencies are now fixed, at least on the\nmaster branch so we can now remove them from the exceptions.\n\nThis change I70ab87c9bed093cad883b6301b8a09753fc470d9 was merged so the\nyaml bug we found in the generator is now fixed across all the branches.\nIt\u0027s now safe to remove the builtin patch.\n\nAlso, os_vif now supports list_opts\n(I7fc132f917949d147dd6371d54e01a6cfd995aae) so we can use it to template\nthe vif_plug_(ovs|linux_bridge) sections. As opposed to cinder, there\u0027s\nno reference section so we need to statically define them.\n\nChange-Id: I6f86ce6b135be9841b5fcae1f940c829327a2db2\n(cherry picked from commit 63f881be6e759e91b008fa25894bd07e1403dbf3)\n(cherry picked from commit d91ed58a6015acdf509bd7d8bbc0287f5a314ca8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/03c3612a89fd0fdf45dd41f4fc16f2e68d11bb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/03c3612a89fd0fdf45dd41f4fc16f2e68d11bb22"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"d925c56f8dd36b3b814d2b4cca9faac83a7db47f","is_merged_in_target_branch":true,"change_id":"I811ee8c811fbc4171c4f54cf8d03e801eec62009","change_number":813096,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"},"3586f730a9dbd4fde791d64d88baaad131eabc5a":{"kind":"REWORK","_number":2,"created":"2021-10-13 15:08:05.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/64/813764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/64/813764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/2"}}},"commit":{"parents":[{"commit":"d925c56f8dd36b3b814d2b4cca9faac83a7db47f","subject":"Rename system_encoding playbooks.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d925c56f8dd36b3b814d2b4cca9faac83a7db47f"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-05 02:35:35.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-13 15:06:57.000000000","tz":-240},"subject":"partial-backport: oslo-config-validator: Removing merged patches","message":"partial-backport: oslo-config-validator: Removing merged patches\n\nWhile developing this, we discovered a lot of inconsistencies across the\nopenstack projects. These inconsistencies are now fixed, at least on the\nmaster branch so we can now remove them from the exceptions.\n\nThis change I70ab87c9bed093cad883b6301b8a09753fc470d9 was merged so the\nyaml bug we found in the generator is now fixed across all the branches.\nIt\u0027s now safe to remove the builtin patch.\n\nPartial-Backport: Some of the changes were not backported before W\n\nChange-Id: I6f86ce6b135be9841b5fcae1f940c829327a2db2\n(cherry picked from commit 63f881be6e759e91b008fa25894bd07e1403dbf3)\n(cherry picked from commit d91ed58a6015acdf509bd7d8bbc0287f5a314ca8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3586f730a9dbd4fde791d64d88baaad131eabc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3586f730a9dbd4fde791d64d88baaad131eabc5a"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"d925c56f8dd36b3b814d2b4cca9faac83a7db47f","is_merged_in_target_branch":true,"change_id":"I811ee8c811fbc4171c4f54cf8d03e801eec62009","change_number":813096,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"},"75b0fc7dced7e0c5811be5b4d9fd60e0835f58a5":{"kind":"REWORK","_number":3,"created":"2021-10-13 15:10:21.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/64/813764/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/64/813764/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/64/813764/3"}}},"commit":{"parents":[{"commit":"ea51007651ba0e13c44f97fcf535a9c84c628b34","subject":"oslo-config-validator: invalidating configs only when necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ea51007651ba0e13c44f97fcf535a9c84c628b34"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-05 02:35:35.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-13 15:10:13.000000000","tz":-240},"subject":"partial-backport: oslo-config-validator: Removing merged patches","message":"partial-backport: oslo-config-validator: Removing merged patches\n\nWhile developing this, we discovered a lot of inconsistencies across the\nopenstack projects. These inconsistencies are now fixed, at least on the\nmaster branch so we can now remove them from the exceptions.\n\nThis change I70ab87c9bed093cad883b6301b8a09753fc470d9 was merged so the\nyaml bug we found in the generator is now fixed across all the branches.\nIt\u0027s now safe to remove the builtin patch.\n\nPartial-Backport: Some of the changes were not backported before W\n\nChange-Id: I6f86ce6b135be9841b5fcae1f940c829327a2db2\n(cherry picked from commit 63f881be6e759e91b008fa25894bd07e1403dbf3)\n(cherry picked from commit d91ed58a6015acdf509bd7d8bbc0287f5a314ca8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/75b0fc7dced7e0c5811be5b4d9fd60e0835f58a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/75b0fc7dced7e0c5811be5b4d9fd60e0835f58a5"}]},"parents_data":[{"branch_name":"refs/heads/stable/victoria","commit_id":"ea51007651ba0e13c44f97fcf535a9c84c628b34","is_merged_in_target_branch":true,"change_id":"Ib3ca74caf747c13937a4efa8c20a6fd789ede43b","change_number":813763,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
