)]}'
{"id":"openstack%2Ftrove-dashboard~1006521","triplet_id":"openstack%2Ftrove-dashboard~stable%2F2026.1~I45193f6f4b9bd0fc446887f2778203e8758d8b8f","project":"openstack/trove-dashboard","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-23 04:13:23.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-23 06:51:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45193f6f4b9bd0fc446887f2778203e8758d8b8f","subject":"Fix empty instance name in delete confirmation on detail page","status":"MERGED","created":"2026-09-21 13:38:01.000000000","updated":"2026-09-23 06:51:13.000000000","submitted":"2026-09-23 06:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006521","cherry_pick_of_change":1006520,"cherry_pick_of_patch_set":1,"meta_rev_id":"de6226a95f14a157dd4d132e3a6086048375bece","_number":1006521,"virtual_id_number":1006521,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 06:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-23 04:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-21 13:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-23 04:13:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 13:38:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-09-21 13:38:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-21 14:01:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61baabf4173d432b5ff583e116aadcd050ccc3fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-21 13:38:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.2.","accounts_in_message":[],"_revision_number":1},{"id":"6af0d13f7fbcb76de6e4b065289ae2c90dcf95e7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-21 13:38:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"88249975d42bb5644bfa69ad031d72c1a54ffa2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:01:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c794377896ba4f28b86ee1c97dbe3794\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28bb904547064dd4a992a3b75170eb22 : SUCCESS in 4m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/70ae64d23b5c4fabbfb014d825c7cd52 : SUCCESS in 4m 52s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9f3f74ef6dc640ba9fafa72a19be4b5f : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c16d408f5824b4994825bd0bd2ae81b : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dc5ff9eb55f412aae2edc2a1a69a91a : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4b7509886f5458a875676d415f11f9d : FAILURE in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64106d45f52e46a1a46b9db4f8a96df4 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c330adea63bb23e6d1b44eddefd5d5b30b3dee57","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-23 04:13:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54dc601c4024506ecb754e4cc65104a493c8722d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:14:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0c988449bee10aebdb362d6018906388bc49a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:07:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4cebb8fec9f4060b59fb3d73fe90e1f\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c537cc7ed3fd4ebbb69c03a2b0fd3434 : SUCCESS in 4m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a693c42aa3d14fabb2924228e97dab79 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba6035fc8ae4a62a58baf88069f0a70 : RETRY_LIMIT in 1m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1839d4bc0b0e458bbe39c2a410ad4876 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed07627125874cecbca140018d1683ee : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"31c3da926cfd45110b8e77350dbf03c63413980f","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-23 06:22:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e13593eca07f6ff8ca36bcc49541a87cae511e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 06:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305c69c8d9eb408893ca252a03798e82\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d8459f2ad94446b89f2e7a6f6151ea4 : SUCCESS in 4m 08s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4939992a7a814628b8df89bd4957bde5 : SUCCESS in 5m 03s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9daaea2d90574f9e9a92013bbae6ad6c : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac71c42776e94b8f859df8e72317519d : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6cff18255cd4b1fb00e670424222cbb : SUCCESS in 7m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2736672c6f38417cabe758b0b0afd684 : FAILURE in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/029220032c7342d5b6bf05e034e34ef8 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a437dd3e2bbed33b7e48c3a7027dfc384479727b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 06:41:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe5de24975f6d5fe68182ec08b0e00861095d113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 06:51:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19fd49b2effc41b183d955ff38243ea5\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b8ecfeedba34894b61c8582bfc82f51 : SUCCESS in 3m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/34962ceebe4c454ab9e703e95960db19 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a9f252de31416cbbaa52f6ce77654e : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd3b31d6dc664ed29cf92e60698d7599 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31274aca293144bbae9b3ddf20be2486 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"de6226a95f14a157dd4d132e3a6086048375bece","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 06:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d54671aa41919a21b1825fc7e7ec6161c7c8c333","revisions":{"d54671aa41919a21b1825fc7e7ec6161c7c8c333":{"kind":"REWORK","_number":1,"created":"2026-09-21 13:38:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/1006521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/21/1006521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/21/1006521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/21/1006521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/21/1006521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/21/1006521/1"}}},"commit":{"parents":[{"commit":"741bd6681c90c6bb12ba06c0c01504e52bd7ca52","subject":"Merge \"Replace usage of oslo.log\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/741bd6681c90c6bb12ba06c0c01504e52bd7ca52"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-21 12:55:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-21 13:38:01.000000000","tz":0},"subject":"Fix empty instance name in delete confirmation on detail page","message":"Fix empty instance name in delete confirmation on detail page\n\nThe Confirm Delete dialog on the instance detail page showed\n\"You have selected: .\" because the overview template did not expose\nthe instance name to Horizon\u0027s JS dialog.\n\nCloses-Bug: #2155808\nChange-Id: I45193f6f4b9bd0fc446887f2778203e8758d8b8f\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d54671aa41919a21b1825fc7e7ec6161c7c8c333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d54671aa41919a21b1825fc7e7ec6161c7c8c333"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"67a22401584391b544b485f893d1f2819803949c","ours":"741bd6681c90c6bb12ba06c0c01504e52bd7ca52","theirs":"67b2ded33eaacce320248af6ea700152808355d6","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
