)]}'
{"id":"openstack%2Ftrove-dashboard~409356","triplet_id":"openstack%2Ftrove-dashboard~master~I0257d1650da870931f4324554f1039f7582af74f","project":"openstack/trove-dashboard","branch":"master","topic":"bug/1648883","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-01 13:20:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-01 12:46:29.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"9664":{"account":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"last_update":"2026-05-25 17:32:21.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"12673":{"account":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"last_update":"2026-09-01 13:08:18.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"24739":{"account":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"last_update":"2026-05-25 17:32:22.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-01 13:20:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"10215":{"account":{"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},"last_update":"2026-05-25 17:32:23.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-09-01 12:03:40.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-09-01 12:03:40.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 12:13:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I0257d1650da870931f4324554f1039f7582af74f","subject":"Add support for upgrade instance","status":"MERGED","created":"2016-12-09 22:34:07.000000000","updated":"2026-09-01 13:32:57.000000000","submitted":"2026-09-01 13:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"409356-bug/1648883","meta_rev_id":"1a4b73eb61c04457ae29dc363c6b3192004b33af","_number":409356,"virtual_id_number":409356,"owner":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},{"value":0,"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 13:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-09-01 12:46:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-09-01 13:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},{"value":0,"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-01 12:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-09-01 13:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},{"value":0,"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-16 21:49:31.000000000","updated_by":{"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},"reviewer":{"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},"state":"REVIEWER"},{"updated":"2017-09-27 01:53:16.000000000","updated_by":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"reviewer":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"state":"REVIEWER"},{"updated":"2017-11-02 19:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-08 21:11:11.000000000","updated_by":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"reviewer":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"state":"REVIEWER"},{"updated":"2018-02-13 03:08:03.000000000","updated_by":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"reviewer":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"state":"REVIEWER"},{"updated":"2026-05-25 17:32:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-25 17:32:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-05-25 21:22:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-05-29 15:06:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2026-08-26 13:54:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"REVIEWER"},{"updated":"2026-08-30 20:27:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-08-31 03:53:09.000000000","updated_by":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"reviewer":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"state":"REMOVED"},{"updated":"2026-08-31 08:46:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"18f316e7649f6989703fb0a6ab100a7f5b18181c","author":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"date":"2016-12-09 22:34:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad2c1fff638ab0ca5d00cdf01ea95b89035be448","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 22:38:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/1/check/gate-trove-dashboard-pep8-ubuntu-xenial/f7b46c2/ : SUCCESS in 2m 52s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/1/check/gate-trove-dashboard-python27-ubuntu-xenial/176b494/ : SUCCESS in 2m 48s\n- gate-trove-dashboard-python34 http://logs.openstack.org/56/409356/1/check/gate-trove-dashboard-python34/8969474/ : FAILURE in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9593ef7779d74d4bc84d1efd14747f968293ed71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-11 17:16:11.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"789fd717b797e315331ed3b5950c9ef510551a1b","author":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"date":"2016-12-12 17:53:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e39cdad3e3fad8eba315fa33791ea0dbe2723af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 18:04:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-pep8-ubuntu-xenial/c4d5286/ : POST_FAILURE in 8m 59s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-python27-ubuntu-xenial/7cbb32f/ : SUCCESS in 2m 46s\n- gate-trove-dashboard-python34 http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-python34/cab00cd/ : FAILURE in 2m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f507149e52458e01bd402fc569988b7cd0e9883a","author":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"date":"2016-12-12 18:05:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6dc07936aba20b15496df477c7b86699f4cad7d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 18:13:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-pep8-ubuntu-xenial/4086c4f/ : SUCCESS in 5m 15s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-python27-ubuntu-xenial/0b22d30/ : SUCCESS in 6m 12s\n- gate-trove-dashboard-python34 http://logs.openstack.org/56/409356/2/check/gate-trove-dashboard-python34/35e143c/ : FAILURE in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57465d4dfca62fb55b6d0b79fb38fee202be93e2","author":{"_account_id":10215,"name":"Peter Stachowski","email":"peter@tesora.com","username":"peterstac"},"date":"2017-01-16 21:49:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d0e4b4bd246415ca7b5861d8539592dba65072b","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-04-26 12:39:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ffe491549966a13ba89f1c8ca8b66e2544378242","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-04-26 12:40:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"559790402dabda2737ef4d2c601edca3aa09711f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-26 12:46:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-trove-dashboard-python35-nv http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-python35-nv/07a0d45/ : FAILURE in 4m 10s (non-voting)\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-pep8-ubuntu-xenial/a7afe64/ : SUCCESS in 5m 04s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-python27-ubuntu-xenial/778c04a/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b694b3c2e84d2d8414c00a47e672521f7172f159","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-26 12:46:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"8e409dc6c936a13aa07a9895c65e6e38599abe77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-26 12:52:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/3/gate/gate-trove-dashboard-pep8-ubuntu-xenial/1ac31f9/ : SUCCESS in 5m 00s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/3/gate/gate-trove-dashboard-python27-ubuntu-xenial/6758f8b/ : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c89fa5a328f1f3d4222360a77eaef8eae869e4ef","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-04-26 13:00:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"521d78a677a23d339cdbb255f902ba3d8e7e724e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-26 13:08:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-dashboard-python35-nv http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-python35-nv/5f03bce/ : FAILURE in 5m 33s (non-voting)\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-pep8-ubuntu-xenial/3a45ad5/ : SUCCESS in 5m 29s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/3/check/gate-trove-dashboard-python27-ubuntu-xenial/978ee05/ : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"61f5fbb5b347f61a290060b696fe826b4204d765","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-06-16 01:38:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2e2e7c745d69ec73c61aff56e56da5448c61d8d2","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-06-16 01:38:05.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"30ac813b6ce8c22d1ff6e7840a9d10b6200d04e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 01:56:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-trove-dashboard-python35-nv http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-python35-nv/d4d5539/ : FAILURE in 6m 17s (non-voting)\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-pep8-ubuntu-xenial/24b6481/ : SUCCESS in 5m 13s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-python27-ubuntu-xenial/97ffe60/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":4},{"id":"6e899e303588d08f513ad656ed987cb36c348ca4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 01:56:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"0a20ddbd232d9a3989b76693aad73b171d80e75d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 02:04:31.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/4/gate/gate-trove-dashboard-pep8-ubuntu-xenial/1aefcbf/ : SUCCESS in 5m 45s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/4/gate/gate-trove-dashboard-python27-ubuntu-xenial/e2554a9/ : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":4},{"id":"36268141cd5082902377686a8e283cae695dd616","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-06-16 10:38:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"63b6b33b6b9e7f98be800dfe5550a6a87221b9bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 10:45:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-dashboard-python35-nv http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-python35-nv/fa01e34/ : FAILURE in 5m 24s (non-voting)\n- gate-trove-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-pep8-ubuntu-xenial/85ac1b1/ : SUCCESS in 5m 22s\n- gate-trove-dashboard-python27-ubuntu-xenial http://logs.openstack.org/56/409356/4/check/gate-trove-dashboard-python27-ubuntu-xenial/fd27945/ : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"7f169be7696ac89cb21aa9361dfbf12dcacdb5d3","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-09-27 01:53:16.000000000","message":"Patch Set 4: -Workflow -Code-Review\n\nremoving reviews","accounts_in_message":[],"_revision_number":4},{"id":"ca5afe60ce64c3ab8e1f09e1b48ed11d8f68e570","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-02 18:51:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a0de2c333f4c667ae9fdddcc0de7bf05dc5c9b88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 19:17:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/409356/4/check/openstack-tox-pep8/9205796/ : SUCCESS in 6m 30s\n- openstack-tox-py27 http://logs.openstack.org/56/409356/4/check/openstack-tox-py27/94f8fb8/ : FAILURE in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/56/409356/4/check/openstack-tox-py35/fa90c73/ : FAILURE in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93b50595ecdcd24b16a5bdca9e393f2454945152","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-08 19:55:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"27abe79a3ca79e81655c2f4a3cc6f3b07d5ec5a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-08 20:06:21.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/409356/4/check/openstack-tox-pep8/04ef213/ : SUCCESS in 9m 29s\n- openstack-tox-py27 http://logs.openstack.org/56/409356/4/check/openstack-tox-py27/15e7839/ : FAILURE in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/56/409356/4/check/openstack-tox-py35/f53bf06/ : FAILURE in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a00e6190dca8e8c75a44e55ee62b8a9de33b726f","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-08 20:21:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4f810fd7cf93d55203acef4042c785c3be6fb861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-08 20:31:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/409356/4/check/openstack-tox-pep8/5670df3/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/56/409356/4/check/openstack-tox-py27/bc724fc/ : FAILURE in 5m 16s\n- openstack-tox-py35 http://logs.openstack.org/56/409356/4/check/openstack-tox-py35/cc2abe6/ : FAILURE in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c310ee7774df8b2a717050d2b12d6a6daaf67d8b","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-08 21:11:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2fb61ed51ec2e86037ed8105291226e4a49c965f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-08 21:16:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/409356/4/check/openstack-tox-pep8/9c16741/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/56/409356/4/check/openstack-tox-py27/124aba4/ : FAILURE in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/56/409356/4/check/openstack-tox-py35/d4ad285/ : FAILURE in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8af93756fd515c1971e63aebd61c4fe59869efd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 01:26:38.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":4},{"id":"f98ff59782e3bd4cbb8d75f9a6d71ab3ae8e6a4a","author":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"date":"2018-02-13 01:45:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"452775498e2b89a0150b42219375998c34283010","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 01:58:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/409356/4/check/openstack-tox-pep8/16caf7d/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/56/409356/4/check/openstack-tox-py27/4073f38/ : FAILURE in 7m 18s\n- openstack-tox-py35 http://logs.openstack.org/56/409356/4/check/openstack-tox-py35/a7375dc/ : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"43478eb2de5521b9b6c4588c49c0ee5eeef44ed3","author":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"date":"2018-02-13 03:08:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77cb9cfa49eadd2c66af6b84c3139605bee63e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-31 22:56:04.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5491c359839fe6185ad5ac74cf3ff49621eaa5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 22:56:24.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove-dashboard for 409356,5","accounts_in_message":[],"_revision_number":5},{"id":"2bbc88cb65cac7ff48e70c0b8f169ee5aa89c5c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-31 22:57:09.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b570035529356d1e5eb7db45721721770f3721d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-31 22:57:12.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9d89e550310dc12b7dec4c85c239dae0a988ff69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-31 22:58:30.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"384a1f7d8980fa11b2e0313aa8f033fd92815351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-31 23:00:36.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"b6c8e5d1d9fee6ec1a3749c4b720e2680025f59d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 23:18:07.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/184d6eb18935489fbb77ce3341768792\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5f1c817bcef457eaadade4e516cfe99 : FAILURE in 3m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/bbc9df239dd546389825009b550c9152 : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95cec8cb49a4eb19f114c71c5178b88 : FAILURE in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc44b63adece4812bce12c200bcd84a0 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":9},{"id":"d51e684192a4dfd3dadd90f0ac2486052de86a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 09:20:28.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"62fea42a504cef91c7c8bc6d237148ee9d19235e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 09:22:01.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"92b6c5ef7e6a80a93b900aa0290731e0c689385f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:29:22.000000000","message":"Patch Set 11: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e696f6b87d5a400381d7e47f66f9a963\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fc62ecacc6b459c8b1a3a9e1380ccd9 : FAILURE in 1m 59s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/767461a400764739a5f81b3ec8ea0895 : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4a1285ed644d58be6a1115346bb3e3 : FAILURE in 1m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f7d1175b520455986b4a796aea6f706 : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":11},{"id":"f824e4480722b922fd3250802abe9db4d4ac32a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 10:13:45.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b8e627dc901f9d4161e612821463e21a60afbd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:22:29.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/637ce4853bb244efa4a0259cf5cf5d3d\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0381e3ae11e14c5f8ada8c45a8780f54 : FAILURE in 3m 30s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4b198106d1c644088f438e60dcc8d2ef : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49f93c41d7a452ebcd01aa8cb49a8aa : FAILURE in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/972233c8334b47bd946930f459d122da : FAILURE in 2m 55s","accounts_in_message":[],"_revision_number":12},{"id":"8a1290cb28c95bdcf66d6776ec5387459423f2e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 17:32:07.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"21cba17e27da47193110eac3d7b73ff903629fb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 17:32:14.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"f110d9635a8270b39c7b9a26c4c7a810b8e31613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-25 17:35:00.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"c0ce8708b6a11bd09ceb7ced85e5a18345e7cf39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 17:46:43.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/964c97c1759a4a708f85141a2cdde9a6\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f67fbaeefb874e6495a4262be36708d7 : FAILURE in 5m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/8fc47ffe504d45aba1a6cc05f092c268 : FAILURE in 2m 44s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/0dd14441da424f44a210f617e65885bd : FAILURE in 4m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c893e382f19a48c0bbecebbea748ec8f : FAILURE in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddb3ba15765b465f9cbb0e0aefe2af1d : FAILURE in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/511bf5ed7003432d9d25bd6a4fe951e2 : FAILURE in 3m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02559bc218274f5b90eb7ce1cbed894c : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":15},{"id":"c705bda3ccf63ecadd331b122febbcc1fd772b69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-25 17:52:08.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"e86240954ed7973466f9083b063741ea54710958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:04:33.000000000","message":"Patch Set 16: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fd5abeb95be42bbb45bf8443da7dfb6\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33ce9b170bf2488d9630627402c23c47 : FAILURE in 2m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/fdebe78a9746462493e362dd16ca4f27 : FAILURE in 4m 35s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/5f491ba7c60f44d7a35b344bc90adf2d : FAILURE in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9097352d52d547b6b777763cb8d0cc72 : FAILURE in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a6c2d0d728644b3ab357597037248d2 : FAILURE in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66e9108efaa8472aa732369f216c1b6b : FAILURE in 6m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca92f878c69f442dbf68c62454eaabae : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":16},{"id":"86edd2504962ce030fa7b91fa62d1dc669058922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-25 18:07:23.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"50fb0f3806df22c2b2790d75f01fef231d2378d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:18:40.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bc8d1e42bef4b1db985cb021bf09b6b\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2bb72cc2d7064372acfbd6a51b71f29f : FAILURE in 3m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/74e5c8b269434348ae38be6f37d6cef9 : FAILURE in 4m 22s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/c85c39c9565442498fcf0f673e750d38 : FAILURE in 4m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b06b5e530c2478fa7551b5819fa0a89 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ee9abcc5c38444f8b544157a261e16d : FAILURE in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b993619e1af84621b93e07acf44a5c4c : FAILURE in 6m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dd3589987e447ee8e98d72912fb0600 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":17},{"id":"98ab3f8747aeb9bd906a4c21865322cfe7f3894b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 18:44:52.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"16c3b3c019cdd4c8eaf5c95697e0dfb0633cfd83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:55:36.000000000","message":"Patch Set 18: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c006e0222dc4a54b20795db22e44cbe\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0524d7a182de4374ace6ddb9f90a6b4b : FAILURE in 3m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/88052b71aa9f452cab4f5a7f1d026427 : FAILURE in 4m 29s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/44da8cddc8b14efeb8fc26f1ced016b1 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a28741d4b04d9ebdb383a07ef9168f : FAILURE in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89427fb02c1e4b478c5e2e84fc0fb7e9 : FAILURE in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d977aed92bd4889a9067ed7c02ac66b : FAILURE in 6m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1558fd305834400782a806cf5030d5b3 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":18},{"id":"bb559ec5802d14bbd4db73e9b4d367e03a3509cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:15:34.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"fcab5514ca5af472e04d68d5e21cd40dd19a9ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:16:00.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"df86f862853b5b92bab9386ad0b699fd1215e287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:17:34.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"1108f135c715cb618a8bd08bda172bfecf3ad3e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:18:51.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"1bfac25b9d8818e3e0542457bae824952c577461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:19:44.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"00c09541741e2f85a4dd5286671966cd217db96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:26:17.000000000","message":"Patch Set 23: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db4b3b25cacc40769c22ae5744c96167\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ce85eaf70124d20bc77f5acef0f9ca6 : FAILURE in 3m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b5667507d5d74f0291a5ccb33f01c9ad : FAILURE in 5m 03s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/54f78c0d94a64f17b0874eea97d7d0de : FAILURE in 4m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc2e99e090146b38d51bd83b39a6ca9 : FAILURE in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28f1f46fb1ae4380ada70e9a45f502b4 : FAILURE in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99e397e9e1c24c1aade2f80651fdd88a : FAILURE in 6m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddb1675a15544adba78d9be34c75b6a8 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":23},{"id":"e6984f4512d23370bad95639e486b938dd5571eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:28:29.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"9afaa7502022d67793f5b7b38e0a9faaa88a6704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:35:25.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"997848039e7f960d66c0b3f9f8f5e86a0a49fd3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:43:30.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"e2e7fc2832288a8e8fd6265fd89be60dc28e3ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:52:40.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc813c6434984f73b9d66544ce73c369\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe7c43e68132450c8525568ede74d6be : FAILURE in 3m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a2a3f161b45941dd9ca213cdd0e8a843 : FAILURE in 6m 10s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/6703e6bbf9394c79ac34a235ce57d3cd : FAILURE in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73b49366f0f4c939c039d9e2f7569e9 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8aba6586c3764ccb87cde09091b4a163 : FAILURE in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70c9854debd145fc84a1eea980de40eb : FAILURE in 5m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c33a77e6b28443f2b028b11413d9c18e : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":26},{"id":"9ef4ba8b4fa94603a0fd406282589fba701ea5f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 19:57:05.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"abc68a36301ed603cb270d6125198eba1be8b9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 20:32:30.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"3b10a70ca5fab61d3b1e0317af138a626ac56e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 20:40:09.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"a426d626babd592a59666c426b0e964fe2c3ae02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 20:52:23.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97e381871bde4d4da1009f44a7f89de7\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/760eb9f0da9546d6820e433648c8381f : FAILURE in 4m 46s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c6f78c38b2b84a4e96fe6a1d6bb75110 : FAILURE in 5m 27s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/8521167a7a45422a93c7ce4e54e6a4ea : FAILURE in 5m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4610596c3446469c8d15315288e0c5ff : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87cb5dc6b80b4442af9f41ac8ea9e4af : FAILURE in 6m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba633b63072a4c98a51f7dece0c589f1 : FAILURE in 8m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97745a65afbe46a29a995f785ab1f55f : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":29},{"id":"ce2ea59eba4f1bca723d9031c1eb44176b91c6c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 20:55:30.000000000","message":"Patch Set 30: Published edit on patch set 29.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"587136ceea7a860f4b6096815000edf06b5c9bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 21:19:57.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3102b21ccc34c3887dfee412579baad\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/779a7027afbc430f86a2d22e29596be9 : SUCCESS in 4m 30s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/8b88d941d01641b1a6ea5758456d0159 : SUCCESS in 6m 53s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/edd39d3d5dcf45f4aeb0b52e4d7568d4 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5548fd977b63490486a170190fe8435e : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c160b1ad3ed435fbc733be9dbc1836e : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/322324fc79b643d097545b4925fc6947 : SUCCESS in 8m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f8be16e1adc42108416d878f28dd082 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":30},{"id":"f70b39ae79cec9ab4bf138a904431b79219b1b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:20:12.000000000","message":"Patch Set 31: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"7401c7d5710ce42e0fcedae783689e411d4a877c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:21:11.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"e342a996f8fe5336f9623d80f9e3a788dd0605e9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:22:19.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"521b180e01498f0a422e49afae12309acdf89bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 21:31:43.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b530e01a785d4d7e860a8f65e3cc196a\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afd45173a6b442188d141dd1bf87bf3a : SUCCESS in 4m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/dce119c936e248c38a6cad356b17430c : SUCCESS in 2m 43s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/41bb46b2c95f4c09bb3492b5ee91850f : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82685c7bebce4b4797b60003c7cb958e : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a205deab16da48e481e72325596436e1 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e326a866b3ba4cad90bcc14e801a240f : SUCCESS in 6m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2749a9aeb438427e8cecf81b4d8275aa : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":31},{"id":"e0b7bcb7cca62fbb30f65ff7e29aefcf47d8ae17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-25 21:39:32.000000000","message":"Patch Set 32: Published edit on patch set 31.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"1aecc9032e14dfb05a858ec609e4021a00cdc8ef","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-25 21:39:37.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"3d5576cdb7011ce892f69f32537a6a4b7438e069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:11:39.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfd91a60742e454e94577bc55d6fa92e\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04ade65433404c4fa30023e8026c6fba : SUCCESS in 4m 31s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/384d65139f4b451a801158794d2d49fd : SUCCESS in 7m 25s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/4a33eaecef4249fdbb2d761466d135c4 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9472ee9e90db4b77873cf7eaa8bada04 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d24a0ea3cb904ebf89e2982b4093bf04 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32a0b4436ce74a66bf8a5874a7a50786 : SUCCESS in 7m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/937fa1fa8b7845ee89bba2f3ed7ae2ed : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":32},{"id":"1173ce212a6684924fde996ba86d27053de942e5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 19:29:33.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"8f7818fc9d1d2fe10d407574434eaa18f21b3eab","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-10 03:47:17.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"5cb948967d925eb287517dc79bb11ab5fb895af2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-18 11:22:15.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"cbcbfe52ade5bf101091092cdec2e3bead378382","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-27 06:00:29.000000000","message":"Patch Set 32:\n\n(4 comments)","accounts_in_message":[],"_revision_number":32},{"id":"473b27c73b00ebdf6ba266796f60844293021054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 07:56:16.000000000","message":"Patch Set 33: Patch Set 32 was rebased on behalf of \u003cGERRIT_ACCOUNT_37632\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"_revision_number":33},{"id":"2adfa0974282c4e75456d82fd4e39f8f0cad0279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 07:56:32.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"f5af24e58e45f02f0241c6270c5aadf3f27cfe6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:00:40.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"47817d94825060c0b366da041c6f96cec83fcfa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:04:19.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"c9fdda90b3bc836a7e1462942422d80aecf04dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:11:22.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"a6b5b5b9c62cb967f0f2c75df9446e72d3990e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:17:49.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"7e9ac650b7ed5abdbf5ff8a9e9d78b5d4728b394","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:29:21.000000000","message":"Patch Set 38:\n\n(4 comments)","accounts_in_message":[],"_revision_number":38},{"id":"3f5e4fc07bf1bbf4c83077e763595e4b8f018da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:36:52.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6501d05c089b4c6d96f2ecdb2e703aee\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/20f40d8edaff49889e8c8008784a8205 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7771e5e8fad44dc89a1245924a24311 : SUCCESS in 4m 02s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/89adeb57c12c43728fa006c49c47b097 : SUCCESS in 7m 10s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ada51ca580094d17bce40318bc468563 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6158f6d8304144bbbeeef5822f5594bd : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d5e2e37ba514f1aacc5015c39721fad : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e53ee4f19874d96a0e6d3ab287bd078 : SUCCESS in 10m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc0aad440e054372819496ec3f5af0d0 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":38},{"id":"420b2a9fe53f826ea222df8d78b971499da1ab5f","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-28 06:18:11.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"83ca2e23c11a43f0dea865512cc83ecadb5f427c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 20:25:01.000000000","message":"Patch Set 39: Published edit on patch set 38.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":39},{"id":"ca61727ca0c45f807afb5ed35a15ccf6c9397bbf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 20:26:02.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"64baf614c65901fd3e8956784d05481276d1c7eb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 20:27:51.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"47cb1ace0ba82dd0bc70f141b60afe2f55e582cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 20:34:10.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"b2c5b6e4382bf6d114dd22e57f574a4c320b6e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 20:50:52.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f021cab22a054807af97b5e3bd50c122\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4b3dc084e70f46fd91380cf123ec0d44 : SUCCESS in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6545593fd80541da9e464de673ece090 : SUCCESS in 6m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/84a8324dff6149e6a5b343ac64342160 : SUCCESS in 4m 31s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/926fc9b79178421594c10978be8b1cf3 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29fbc13e47047cba5f7215e93ecbc6d : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f1b4535ecb84051a6d479f7601e3195 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4797cd95e7c47919ce9c7f50c414519 : SUCCESS in 9m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e562248b4b48998d15e364ddb657d8 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":40},{"id":"4386ffd70ec35d6aa60c81f1d8a1ed5723ea76f3","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-31 03:34:40.000000000","message":"Patch Set 40: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"22c2b9e6fbfdad2d8f1be7af751575f65f8e0b99","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2026-08-31 03:53:09.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_9664\u003e.","accounts_in_message":[{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"}],"_revision_number":40},{"id":"cd9e1ac9735859ea9168fffe27b18d0d85f2d507","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 08:46:14.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"7b53e637769b316273390353e630e5bba9aeb7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-01 11:58:50.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"12dde996ab0c20ec2a7188de5b61eff0711c28a0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-01 12:03:40.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"321d36fce032773a6dd8faa227199e9ee888736a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 12:13:02.000000000","message":"Patch Set 41: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"6dab35ead45d121fcae3f27744cf4ebe6a48d1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:20:49.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668de2caed40492ca3c3f6e7a65600b1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6617304000264ab98baeb224079d75b4 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21ff7a68cb884ae0b8ce537c8c829926 : SUCCESS in 5m 51s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6d92715224494e838d93e6b85b5741ac : SUCCESS in 3m 53s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/0c8bc19307094fefb2da715ce73055f6 : SUCCESS in 5m 00s\n- horizon-tox-python3-django60 https://zuul.opendev.org/t/openstack/build/aa7d11ae11384c9184c7ef183c0d6159 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f49d17f474d4d389b95bfd94c18af48 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/214432dbef8a42e2a6604d306e8d5b8a : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0fca961bc13245afa96105b07309daea : SUCCESS in 7m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9947289ed27048e0aa839c8e28653520 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":41},{"id":"4c42de0a60230f4347aaeb048cf3831a7c9eab09","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-01 12:46:29.000000000","message":"Patch Set 41: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"9e25cb34683e2a32f0a6248fcc47f02a1c14f55b","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-01 13:20:28.000000000","message":"Patch Set 41: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"036b7a35f1795c00a3a32529a96a03201a731cf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:20:39.000000000","message":"Patch Set 41: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":41},{"id":"c1114b22ce92c3e155c8e4ef42c6aec3c4cee49e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:31:59.000000000","message":"Patch Set 41: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d17c17a4ad47818bbc26cd9dbe0d97\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/78d4d17e8cf543c685a73f500b912d2f : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d007651c4af4716b8ecfbcf4d0f29d6 : SUCCESS in 4m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5e5d2c2336494097b55b9c0fd25745b5 : SUCCESS in 4m 50s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/14e2999dbbe143edb0376e6ece65b3b4 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c46b78c7ab54a16b201be9621c4cfc3 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5ea6fc11b234442a79b59338e8cd354 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ca058a729d7416bb5bfcd533ec5c524 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":41},{"id":"51fd1ca90811c9037fa76d52701115c11fc08876","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:32:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":41},{"id":"1a4b73eb61c04457ae29dc363c6b3192004b33af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:32:57.000000000","message":"Patch Set 41:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29e1ad0af454057965e63e06d939b58\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ac217219c94062b039b27e778da439 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"806f4b5514a4bd0f07a0d87cb1ef3b6257fb3986","revisions":{"54adc93f791e16f2de644643481d9b0d05663ea0":{"kind":"REWORK","_number":1,"created":"2016-12-09 22:34:07.000000000","uploader":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"ref":"refs/changes/56/409356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/1"}}},"commit":{"parents":[{"commit":"230a112ee6f5d11f31d4cf89cc7ffa9e3fd0594a","subject":"Add requirement for cross-process cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/230a112ee6f5d11f31d4cf89cc7ffa9e3fd0594a"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-12-09 22:28:29.000000000","tz":-300},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/54adc93f791e16f2de644643481d9b0d05663ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/54adc93f791e16f2de644643481d9b0d05663ea0"}]},"branch":"refs/heads/master"},"fab8e88f616fd8b1c906d6b91186dafc774e2028":{"kind":"REWORK","_number":2,"created":"2016-12-12 17:53:52.000000000","uploader":{"_account_id":12673,"name":"Duk Loi","email":"duk@tesora.com","username":"duktesora"},"ref":"refs/changes/56/409356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/2"}}},"commit":{"parents":[{"commit":"2e7ad6c53570265ec66d8e585c2c504e9d23a279","subject":"Merge \"Fix Django 1.10 Compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/2e7ad6c53570265ec66d8e585c2c504e9d23a279"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-12-12 15:28:44.000000000","tz":-300},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/fab8e88f616fd8b1c906d6b91186dafc774e2028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/fab8e88f616fd8b1c906d6b91186dafc774e2028"}]},"branch":"refs/heads/master"},"46626f5a9fb18ceefcd58de54656a684fd03d4d4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-04-26 12:39:56.000000000","uploader":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"ref":"refs/changes/56/409356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/3"}}},"commit":{"parents":[{"commit":"bbb52c7e4e1a24aaf9bcc373df6449cd92145313","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bbb52c7e4e1a24aaf9bcc373df6449cd92145313"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Amrith Kumar","email":"amrith.kumar@gmail.com","date":"2017-04-26 12:39:56.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46626f5a9fb18ceefcd58de54656a684fd03d4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46626f5a9fb18ceefcd58de54656a684fd03d4d4"}]},"branch":"refs/heads/master"},"af13e5ff3b4c30f29b78b81be9fc6521440e1eb4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-06-16 01:38:00.000000000","uploader":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"ref":"refs/changes/56/409356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/4"}}},"commit":{"parents":[{"commit":"76833761edeb310f136102a4640dba491d3ae2bb","subject":"Merge \"Update launchpad link to trove-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/76833761edeb310f136102a4640dba491d3ae2bb"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Amrith Kumar","email":"amrith.kumar@gmail.com","date":"2017-06-16 01:38:00.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/af13e5ff3b4c30f29b78b81be9fc6521440e1eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/af13e5ff3b4c30f29b78b81be9fc6521440e1eb4"}]},"branch":"refs/heads/master"},"679bbfe576359cee0611d47a54448f716c9ca0f5":{"kind":"REWORK","_number":5,"created":"2025-03-31 22:56:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/5"}}},"commit":{"parents":[{"commit":"76833761edeb310f136102a4640dba491d3ae2bb","subject":"Merge \"Update launchpad link to trove-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/76833761edeb310f136102a4640dba491d3ae2bb"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-03-31 22:56:04.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/679bbfe576359cee0611d47a54448f716c9ca0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/679bbfe576359cee0611d47a54448f716c9ca0f5"}]},"branch":"refs/heads/master"},"2f6f860b75be9b3b5ecbf0f7f05b401c457bac29":{"kind":"REWORK","_number":6,"created":"2025-03-31 22:57:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/6"}}},"commit":{"parents":[{"commit":"76833761edeb310f136102a4640dba491d3ae2bb","subject":"Merge \"Update launchpad link to trove-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/76833761edeb310f136102a4640dba491d3ae2bb"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-03-31 22:57:05.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/2f6f860b75be9b3b5ecbf0f7f05b401c457bac29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/2f6f860b75be9b3b5ecbf0f7f05b401c457bac29"}]},"branch":"refs/heads/master"},"deb828fc8a3045d704da5fe1d11b95e0918abb26":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-03-31 22:57:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/7"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-03-31 22:57:12.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/deb828fc8a3045d704da5fe1d11b95e0918abb26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/deb828fc8a3045d704da5fe1d11b95e0918abb26"}]},"branch":"refs/heads/master","description":"Rebase"},"56ddc4f6973755bf8c7f8844b1bb8588aa634f40":{"kind":"REWORK","_number":8,"created":"2025-03-31 22:58:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/8"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-03-31 22:58:30.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/56ddc4f6973755bf8c7f8844b1bb8588aa634f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/56ddc4f6973755bf8c7f8844b1bb8588aa634f40"}]},"branch":"refs/heads/master"},"a1cd8bcdbe35dec470d07b03dd067d9e54d5c89b":{"kind":"REWORK","_number":9,"created":"2025-03-31 23:00:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/9"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-03-31 23:00:30.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a1cd8bcdbe35dec470d07b03dd067d9e54d5c89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a1cd8bcdbe35dec470d07b03dd067d9e54d5c89b"}]},"branch":"refs/heads/master"},"490c4886bfd2d30ae2071022e565ccf900946ee0":{"kind":"REWORK","_number":10,"created":"2025-04-01 09:20:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/10"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-01 09:20:28.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/490c4886bfd2d30ae2071022e565ccf900946ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/490c4886bfd2d30ae2071022e565ccf900946ee0"}]},"branch":"refs/heads/master"},"51fb5556e7157062fab43ef36fc50aa0e3ada738":{"kind":"REWORK","_number":11,"created":"2025-04-01 09:22:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/11"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-01 09:22:01.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/51fb5556e7157062fab43ef36fc50aa0e3ada738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/51fb5556e7157062fab43ef36fc50aa0e3ada738"}]},"branch":"refs/heads/master"},"0a87cd658cf9b8dc090c970345eac17978ee0773":{"kind":"REWORK","_number":12,"created":"2025-04-01 10:13:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/12"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-01 10:13:45.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCloses-Bug: #1648883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0a87cd658cf9b8dc090c970345eac17978ee0773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0a87cd658cf9b8dc090c970345eac17978ee0773"}]},"branch":"refs/heads/master"},"5fa1276efc431c60c8a5ae77530b1a7e3dcc407c":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-05-25 17:32:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/13"}}},"commit":{"parents":[{"commit":"ae6af4ec181a055a9eaebb2fd747e1ac99418129","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ae6af4ec181a055a9eaebb2fd747e1ac99418129"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 17:32:07.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/5fa1276efc431c60c8a5ae77530b1a7e3dcc407c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/5fa1276efc431c60c8a5ae77530b1a7e3dcc407c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e6b20e8d01cd2d5a3c9a15ab0120a47e46fde523":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-05-25 17:32:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/14"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 17:32:14.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/e6b20e8d01cd2d5a3c9a15ab0120a47e46fde523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/e6b20e8d01cd2d5a3c9a15ab0120a47e46fde523"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"5fa1276efc431c60c8a5ae77530b1a7e3dcc407c","theirs":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","contains_conflicts":false}},"0559254fa2cecb49600f50b3ba037a5ad958a2ed":{"kind":"REWORK","_number":15,"created":"2026-05-25 17:35:00.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/409356/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/15"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-05-25 17:34:44.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0559254fa2cecb49600f50b3ba037a5ad958a2ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0559254fa2cecb49600f50b3ba037a5ad958a2ed"}]},"branch":"refs/heads/master"},"0f2c40dd21adbfd3255e4b7d21e486920007d839":{"kind":"REWORK","_number":16,"created":"2026-05-25 17:52:08.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/409356/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/16"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-05-25 17:51:57.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0f2c40dd21adbfd3255e4b7d21e486920007d839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0f2c40dd21adbfd3255e4b7d21e486920007d839"}]},"branch":"refs/heads/master"},"8a40bd992606f57b73228a2f1c7a76f7466598ed":{"kind":"REWORK","_number":17,"created":"2026-05-25 18:07:23.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/409356/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/17"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-05-25 18:07:17.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8a40bd992606f57b73228a2f1c7a76f7466598ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8a40bd992606f57b73228a2f1c7a76f7466598ed"}]},"branch":"refs/heads/master"},"59d0b1811f8fbb1962d26e776f61515f2ce139be":{"kind":"REWORK","_number":18,"created":"2026-05-25 18:44:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/18"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 18:44:51.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/59d0b1811f8fbb1962d26e776f61515f2ce139be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/59d0b1811f8fbb1962d26e776f61515f2ce139be"}]},"branch":"refs/heads/master"},"47f200864d90a4f58133044a25875b6b4001ce2a":{"kind":"REWORK","_number":19,"created":"2026-05-25 19:15:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/19"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:15:33.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/47f200864d90a4f58133044a25875b6b4001ce2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/47f200864d90a4f58133044a25875b6b4001ce2a"}]},"branch":"refs/heads/master"},"0047ccd183e2580f2348caae0043d20264eba416":{"kind":"REWORK","_number":20,"created":"2026-05-25 19:16:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/20"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:16:00.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0047ccd183e2580f2348caae0043d20264eba416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0047ccd183e2580f2348caae0043d20264eba416"}]},"branch":"refs/heads/master"},"ec0a0c189ce839ca3efeb62c7ad1c5a42626010c":{"kind":"REWORK","_number":21,"created":"2026-05-25 19:17:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/21"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:17:34.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ec0a0c189ce839ca3efeb62c7ad1c5a42626010c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ec0a0c189ce839ca3efeb62c7ad1c5a42626010c"}]},"branch":"refs/heads/master"},"b3485b3a5ed8fc3934f324b258bb9d26cf30cc25":{"kind":"REWORK","_number":22,"created":"2026-05-25 19:18:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/22"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:18:51.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b3485b3a5ed8fc3934f324b258bb9d26cf30cc25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b3485b3a5ed8fc3934f324b258bb9d26cf30cc25"}]},"branch":"refs/heads/master"},"46cfad0bfdb87a0b5cd35a54bbca5f7227b7de39":{"kind":"REWORK","_number":23,"created":"2026-05-25 19:19:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/23"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:19:43.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46cfad0bfdb87a0b5cd35a54bbca5f7227b7de39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46cfad0bfdb87a0b5cd35a54bbca5f7227b7de39"}]},"branch":"refs/heads/master"},"3bb4d984e52ddc105cf145464ae14f7f8a955fb1":{"kind":"REWORK","_number":24,"created":"2026-05-25 19:28:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/24"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:28:29.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/3bb4d984e52ddc105cf145464ae14f7f8a955fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/3bb4d984e52ddc105cf145464ae14f7f8a955fb1"}]},"branch":"refs/heads/master"},"7ab7416da9726b6bf400de4f852e482688b3d122":{"kind":"REWORK","_number":25,"created":"2026-05-25 19:35:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/25"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:35:25.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/7ab7416da9726b6bf400de4f852e482688b3d122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/7ab7416da9726b6bf400de4f852e482688b3d122"}]},"branch":"refs/heads/master"},"447f90cf99f09cf7fda91dc43dd7a944e7a50e9b":{"kind":"REWORK","_number":26,"created":"2026-05-25 19:43:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/26"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:43:30.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/447f90cf99f09cf7fda91dc43dd7a944e7a50e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/447f90cf99f09cf7fda91dc43dd7a944e7a50e9b"}]},"branch":"refs/heads/master"},"a94917c49e0c0ea06a178f3fc48b52aa6167b68d":{"kind":"REWORK","_number":27,"created":"2026-05-25 19:57:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/27"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 19:57:05.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a94917c49e0c0ea06a178f3fc48b52aa6167b68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a94917c49e0c0ea06a178f3fc48b52aa6167b68d"}]},"branch":"refs/heads/master"},"a26a20d2d56f98ea7077b66767c018616410d45d":{"kind":"REWORK","_number":28,"created":"2026-05-25 20:32:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/28"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 20:32:29.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a26a20d2d56f98ea7077b66767c018616410d45d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/a26a20d2d56f98ea7077b66767c018616410d45d"}]},"branch":"refs/heads/master"},"f48d83ce9685e3a53da57bd6c539b71ad75038df":{"kind":"REWORK","_number":29,"created":"2026-05-25 20:40:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/29"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 20:40:09.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/f48d83ce9685e3a53da57bd6c539b71ad75038df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/f48d83ce9685e3a53da57bd6c539b71ad75038df"}]},"branch":"refs/heads/master"},"5a36f88a91fb99931fd0b4dd43391eb61435e103":{"kind":"REWORK","_number":30,"created":"2026-05-25 20:55:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/30"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 20:55:30.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/5a36f88a91fb99931fd0b4dd43391eb61435e103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/5a36f88a91fb99931fd0b4dd43391eb61435e103"}]},"branch":"refs/heads/master"},"12121195b2e86002ea3b6205a04abdd56460875a":{"kind":"NO_CODE_CHANGE","_number":31,"created":"2026-05-25 21:20:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/31"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:20:12.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/12121195b2e86002ea3b6205a04abdd56460875a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/12121195b2e86002ea3b6205a04abdd56460875a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9af2494abae47ae6ece94043cc721a0d56c7b038":{"kind":"REWORK","_number":32,"created":"2026-05-25 21:39:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/409356/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/32"}}},"commit":{"parents":[{"commit":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","subject":"Merge \"Remove unused legacy network templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-05-25 21:37:38.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/9af2494abae47ae6ece94043cc721a0d56c7b038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/9af2494abae47ae6ece94043cc721a0d56c7b038"}]},"branch":"refs/heads/master"},"1ff5e8683214c3ab0874a8329c94bbd7815a64c7":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2026-08-27 07:56:16.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/33"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-27 07:56:16.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/1ff5e8683214c3ab0874a8329c94bbd7815a64c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/1ff5e8683214c3ab0874a8329c94bbd7815a64c7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bdfe7a99e4b16e0b77d8bf9b904f4af92696f0a1","ours":"9af2494abae47ae6ece94043cc721a0d56c7b038","theirs":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","merge_strategy":"recursive","contains_conflicts":false}},"369ce0de725cb44b147d80f07ccaaba2799179bb":{"kind":"REWORK","_number":34,"created":"2026-08-27 07:56:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/34"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 07:56:29.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/369ce0de725cb44b147d80f07ccaaba2799179bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/369ce0de725cb44b147d80f07ccaaba2799179bb"}]},"branch":"refs/heads/master"},"104b154d3c204b6aa2461fcf543bb70f80f3468d":{"kind":"REWORK","_number":35,"created":"2026-08-27 08:00:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/35"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 08:00:37.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/104b154d3c204b6aa2461fcf543bb70f80f3468d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/104b154d3c204b6aa2461fcf543bb70f80f3468d"}]},"branch":"refs/heads/master"},"9f00e9e5c27d1db7812f11b25b27205dd3cc9879":{"kind":"REWORK","_number":36,"created":"2026-08-27 08:04:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/36"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 08:04:19.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/9f00e9e5c27d1db7812f11b25b27205dd3cc9879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/9f00e9e5c27d1db7812f11b25b27205dd3cc9879"}]},"branch":"refs/heads/master"},"8f871cba34d8058e095ff7b32f1bcbccecce26a1":{"kind":"REWORK","_number":37,"created":"2026-08-27 08:11:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/37"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 08:11:22.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8f871cba34d8058e095ff7b32f1bcbccecce26a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8f871cba34d8058e095ff7b32f1bcbccecce26a1"}]},"branch":"refs/heads/master"},"ea0c19f6383630db91951c49e7d6e4a57d6db97d":{"kind":"REWORK","_number":38,"created":"2026-08-27 08:17:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/38"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 08:17:49.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ea0c19f6383630db91951c49e7d6e4a57d6db97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ea0c19f6383630db91951c49e7d6e4a57d6db97d"}]},"branch":"refs/heads/master"},"d8b2ee400c1a0bf99a2eb95f9729cf085716b4be":{"kind":"REWORK","_number":39,"created":"2026-08-30 20:25:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/39"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 20:25:01.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d8b2ee400c1a0bf99a2eb95f9729cf085716b4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d8b2ee400c1a0bf99a2eb95f9729cf085716b4be"}]},"branch":"refs/heads/master"},"69f636907c8b3d24d89862e4ace28a08d79d5d87":{"kind":"REWORK","_number":40,"created":"2026-08-30 20:34:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/409356/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/40"}}},"commit":{"parents":[{"commit":"4b5e6bf076811d79de7e09e72be84c6383a6afdf","subject":"Merge \"Drop unused python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/4b5e6bf076811d79de7e09e72be84c6383a6afdf"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 20:34:09.000000000","tz":0},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions (with the\ncurrent version filterd out of the list).\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/69f636907c8b3d24d89862e4ace28a08d79d5d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/69f636907c8b3d24d89862e4ace28a08d79d5d87"}]},"branch":"refs/heads/master"},"806f4b5514a4bd0f07a0d87cb1ef3b6257fb3986":{"kind":"REWORK","_number":41,"created":"2026-09-01 11:58:50.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/409356/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/56/409356/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/56/409356/41"}}},"commit":{"parents":[{"commit":"6afd1a47c3ff5abda3e468680444b11f2f42536f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/6afd1a47c3ff5abda3e468680444b11f2f42536f"}]}],"author":{"name":"Duk Loi","email":"duk@tesora.com","date":"2016-03-31 16:17:51.000000000","tz":-240},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-01 11:58:47.000000000","tz":300},"subject":"Add support for upgrade instance","message":"Add support for upgrade instance\n\nAdded a row menu action for upgrade instance only if there are\nmultiple datastore versions for the instance datastore.\n\nThe action will bring up a dialog showing the current version\nand a dropdown listing all the datastore versions.\n\nAdded a new trove api for instance_upgrade.\n\nCloses-Bug: #1648883\nChange-Id: I0257d1650da870931f4324554f1039f7582af74f\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nCo-authored-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nCo-authored-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\nSigned-off-by: Duk Loi \u003cduk@tesora.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/806f4b5514a4bd0f07a0d87cb1ef3b6257fb3986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/806f4b5514a4bd0f07a0d87cb1ef3b6257fb3986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
