)]}'
{"id":"openstack%2Ftrove-dashboard~680593","triplet_id":"openstack%2Ftrove-dashboard~master~Ic3e8ba5dfa2a91a22913a01facf0e88784583618","project":"openstack/trove-dashboard","branch":"master","topic":"refactor-error-messages","hashtags":[],"change_id":"Ic3e8ba5dfa2a91a22913a01facf0e88784583618","subject":"Refactor error messages","status":"NEW","created":"2019-09-06 07:37:21.000000000","updated":"2020-09-16 10:03:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0cabb95552cb8afe4f1503a79856facd457d9039","_number":680593,"virtual_id_number":680593,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-12 07:35:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-2,"date":"2020-09-16 10:03:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2019-10-13 08:58:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 13:31:02.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-10-12 07:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-13 08:58:23.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2020-09-16 10:03:34.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"42125821eb26e20068d8fd73e26a3a7e6af7e0f2","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-06 07:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4241312ab66f1408bbb8362cd05f5144d83e034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 07:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f6608489de40b28788bf63fadca4b7 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c0a372117834ddabdf989b513e3622a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d7c7a4882bb4b4193513d9bd3c90d5c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd467b048029452f90027aee6276f887 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8d12d0d3ec0e79ef4fe2e573a57d51e88a61bac9","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-09-11 15:45:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a72821c62334171d9dda62246b849680272f78d6","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-20 09:01:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c99ea2d76493c4a71c663d760faed1e55946da0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-20 09:18:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"822b178c9cdf455eb567242b26ff9c5c1786a498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:30:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a0847b718742c0abeb8977a334f2e6 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/120f33c142894723a29c1efcbb9554e9 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fad8c2d25de4272a9642ce6ac5fe061 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91c0e972dfc34a219ac9469ccbfa5d92 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"26306b1fa308c2eec5de148b12c285367259c8b4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-21 14:12:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"997a7f35d876c1f012be61176ae1497b0978e400","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-23 01:32:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"74171d0ff71cf8ce640097314581f51c08fa267c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 01:42:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a932b9a2d23b45b7a320e2efbe0a8b2c : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a6a89204cc1437ea4f3d944c840d2e3 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8a9551dad2e4cc59b832e3edd2f0bb5 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83f13e51cbe04cd29acf028a2ab971ad : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2c8bd67022fba5ffe04b06c2880deed17817360f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-23 06:07:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nKindly also check remaining files as well.","accounts_in_message":[],"_revision_number":3},{"id":"b4f2a60b998cb226cf8077a4dfaf5071aac598f0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-08 07:28:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02283b4ffa5b5fa1ffd94780a7f2b66e85ccc578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 07:39:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76eb61924ae45ea907120fad8e1f768 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b4f4af4fd554fd1b7192e53bc28e057 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cfd9002806d4122ac0559dcdc779c65 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50535441ab384dffaf53064795434e75 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"143a60086f46e370a1ddbbd2cabc9a86d3050cad","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-10-10 06:56:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d8bb9f8f6fed96ed953fa7af3dacc02079b76423","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-10 13:31:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b0b478001299a72064f3f5ec752fc3757070e406","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-12 06:08:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c509b5c2fb117ec58866cb887c7b2e97fbf26b53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 06:22:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d96a17f50db4231a106f1c035bfc5c3 : FAILURE in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92730fc309cf486eb9b2341193ca7b07 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/024edfa7e483456596715bf6faaeaf19 : FAILURE in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b0c1c54b8ea46c98bf7a21684071c97 : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"a22983c2632c9e255aaf0b15d315c1a5ce82a930","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-12 07:19:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7af9668e78d05decdb147e10f23f662fcf0101d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 07:35:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd0bbdfc73b403780e2b1028285f0f5 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ad63fb47524de2b243ca0da302d806 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8c7222107144b19650cd47b459f3e7 : SUCCESS in 11m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56708fa518564aefb566663c1062aa4e : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":6},{"id":"bfbfe555114e402c353b54775562987a4f088f7b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-10-13 08:58:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"3cdacd2258df1765d6800a32760fdf54c4ee85b0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-16 10:03:34.000000000","message":"Patch Set 6: Code-Review-2\n\nThis change is no longer required as it fixed by https://review.opendev.org/#/c/708069/.\nCould you please abandon it.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"66c51e58580378d450cfce48d5ad9cdd251a811d","revisions":{"8052824279bd12c5ad5a3cb62172c31e3afd7912":{"kind":"REWORK","_number":1,"created":"2019-09-06 07:37:21.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/1"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:39.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patchrefactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8052824279bd12c5ad5a3cb62172c31e3afd7912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8052824279bd12c5ad5a3cb62172c31e3afd7912"}]},"branch":"refs/heads/master"},"87dd208c8a64025ee4a7515199ed30a7d3b2e6ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-20 09:18:32.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/2"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-20 09:18:52.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patch refactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/87dd208c8a64025ee4a7515199ed30a7d3b2e6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/87dd208c8a64025ee4a7515199ed30a7d3b2e6ea"}]},"branch":"refs/heads/master"},"fcf96e42be9469e7bec98a9a7485fbc6264cb25d":{"kind":"REWORK","_number":3,"created":"2019-09-23 01:32:15.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/3"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-23 01:32:41.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patch refactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/fcf96e42be9469e7bec98a9a7485fbc6264cb25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/fcf96e42be9469e7bec98a9a7485fbc6264cb25d"}]},"branch":"refs/heads/master"},"45f6cfe44245f8f5825e40da28b7abcdc5d04ee3":{"kind":"REWORK","_number":4,"created":"2019-10-08 07:28:10.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/4"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-08 07:28:24.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patch refactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/45f6cfe44245f8f5825e40da28b7abcdc5d04ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/45f6cfe44245f8f5825e40da28b7abcdc5d04ee3"}]},"branch":"refs/heads/master"},"8b5c161ae2f2b22c4656b82a0ca46196393cbf99":{"kind":"REWORK","_number":5,"created":"2019-10-12 06:08:02.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/5"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-12 06:08:28.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patch refactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8b5c161ae2f2b22c4656b82a0ca46196393cbf99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8b5c161ae2f2b22c4656b82a0ca46196393cbf99"}]},"branch":"refs/heads/master"},"66c51e58580378d450cfce48d5ad9cdd251a811d":{"kind":"REWORK","_number":6,"created":"2019-10-12 07:19:31.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/93/680593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/93/680593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/93/680593/6"}}},"commit":{"parents":[{"commit":"b2df5ef68943d3682a7c562e0b402efe3b020f05","subject":"Update tox.ini for Python 3 Train unit test goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b2df5ef68943d3682a7c562e0b402efe3b020f05"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 07:37:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-12 07:19:47.000000000","tz":480},"subject":"Refactor error messages","message":"Refactor error messages\n\nThis patch refactoring the error message and\nshows the detailed error message\n\nChange-Id: Ic3e8ba5dfa2a91a22913a01facf0e88784583618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/66c51e58580378d450cfce48d5ad9cdd251a811d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/66c51e58580378d450cfce48d5ad9cdd251a811d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
