)]}'
{"id":"openstack%2Ftrove-dashboard~754143","triplet_id":"openstack%2Ftrove-dashboard~master~I99409c9081bdf15171ed9c7309be763d76793cc1","project":"openstack/trove-dashboard","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I99409c9081bdf15171ed9c7309be763d76793cc1","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-24 16:24:25.000000000","updated":"2020-09-24 19:56:18.000000000","submitted":"2020-09-24 19:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754143-1600977378432-60701959","meta_rev_id":"4ccd8de4ea90fbb5dacc0c98fcb28effdfb024d5","_number":754143,"virtual_id_number":754143,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 19:56:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-24 19:41:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-24 19:41:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 19:41:30.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-09-24 19:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57d721002506fd71e74d7a8c6a5db679573b1ba3","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-24 16:24:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3682ae0059967be0b8a6c05dca1beeccc63ee185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 17:05:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3bccdf3861f4a3e9018644f50de24fb : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb482a8ad57472cae4e7f457f7f9e89 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e0b84306104498189f215ae078b344a : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9279f56d8367d2cbb63c480742ec4e8eea661631","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 19:41:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dab360ad0163d7087f277f8563a11abf3e4ccfc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 19:46:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6145be7d8c99e9ee05acf096b83f6b3367e525d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 19:56:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a07d219e606a8ddc7a6add20563d56d0ac760dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 19:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/536db61e52294104af550679b0bad192 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec70f9712e9433dab425037cf31c866 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/966f788ac1534d199a0b38453db8ab94 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"018b3545bcc419e33aeb3963d3fda9247d2d9e57","revisions":{"018b3545bcc419e33aeb3963d3fda9247d2d9e57":{"kind":"REWORK","_number":1,"created":"2020-09-24 16:24:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/43/754143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/43/754143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/43/754143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/43/754143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/43/754143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/43/754143/1"}}},"commit":{"parents":[{"commit":"918379123ba52e61a18f39274be245c5188dfd74","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/918379123ba52e61a18f39274be245c5188dfd74"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 16:24:23.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 16:24:23.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I99409c9081bdf15171ed9c7309be763d76793cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/018b3545bcc419e33aeb3963d3fda9247d2d9e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/018b3545bcc419e33aeb3963d3fda9247d2d9e57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
