)]}'
{"id":"openstack%2Ftrove-dashboard~899133","triplet_id":"openstack%2Ftrove-dashboard~master~Id35b91570515515f39bf54b378cf75f2b20ff8cc","project":"openstack/trove-dashboard","branch":"master","hashtags":[],"change_id":"Id35b91570515515f39bf54b378cf75f2b20ff8cc","subject":"[CI] fix build-openstack-releasenotes job","status":"MERGED","created":"2023-10-24 08:48:25.000000000","updated":"2023-10-24 09:24:37.000000000","submitted":"2023-10-24 09:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899133","meta_rev_id":"ac1c1509de8e9be8609c5f3cc26ae002fc4d96dd","_number":899133,"virtual_id_number":899133,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-24 09:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-24 09:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-24 09:19:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-24 08:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb1305b4f29d27c35f634e8c3e1968b3509ca2a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 08:48:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17d56c52215ad96eda9eb935972711b5c48cb104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 08:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6e9af4500f4a8bb2aad23fc296cc29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d1ca54df73464a971f1713f9d22f9d : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39079dfe7d204eb283b489570d65161f : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c8643c0a90ed3486bd9ca408823e0b35b8cb441f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 09:19:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf604dcad1a904e556c48c4fc67d8689525eda19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 09:19:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00ab6e95a0a72e22710c33968eeef5eb3485f7e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 09:23:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f610b9b947e34b3aaf560070c8ff4dfd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ad596787874da2ab2a951d163eba8c : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2cd77b6a22045a5b959e4a7b2f46a8a : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6bdf6380c303d99086e6c0f85ad8441fddb812a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 09:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac1c1509de8e9be8609c5f3cc26ae002fc4d96dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 09:24:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1823a07a081b4cdb86a72e8ca3af0ae7\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ccff5919b3a45b989e1aae1b870ddda : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1591a734487805e613c2b745864c5926262a6c1","revisions":{"c1591a734487805e613c2b745864c5926262a6c1":{"kind":"REWORK","_number":1,"created":"2023-10-24 08:48:25.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/33/899133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/33/899133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/33/899133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/33/899133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/33/899133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/33/899133/1"}}},"commit":{"parents":[{"commit":"6bc43e214846d60ae77461728033d9f6e4e6feb5","subject":"Add python 3.10 to setup.cfg metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/6bc43e214846d60ae77461728033d9f6e4e6feb5"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-24 08:47:17.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-24 08:47:17.000000000","tz":480},"subject":"[CI] fix build-openstack-releasenotes job","message":"[CI] fix build-openstack-releasenotes job\n\nThe origin error message: Invalid configuration value found:\n\u0027language \u003d None\u0027.\n\nChange-Id: Id35b91570515515f39bf54b378cf75f2b20ff8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/c1591a734487805e613c2b745864c5926262a6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/c1591a734487805e613c2b745864c5926262a6c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
