)]}'
{"id":"openstack%2Ftrove-dashboard~950802","triplet_id":"openstack%2Ftrove-dashboard~master~I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45","project":"openstack/trove-dashboard","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-07-04 16:05:08.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2025-07-04 15:56:33.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-07-04 16:05:08.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-04 16:01:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45","subject":"Remove unicode prefix from code","status":"MERGED","created":"2025-05-23 21:40:42.000000000","updated":"2025-07-04 16:05:08.000000000","submitted":"2025-07-04 16:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950802","meta_rev_id":"260782d7cc04b33ca826c3c0fcd826bbe9a060c0","_number":950802,"virtual_id_number":950802,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-04 16:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-07-04 15:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-07-04 15:56:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 21:51:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-23 21:55:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2025-05-23 21:55:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-05-23 21:55:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"c5d72df90313ec65832c05642b99693502a9b586","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:40:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb19ab9f35be848fa71a5302c35d957d2e4d6d1d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:45:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6efad80642e1131247cb3b5634b7b5d3c80205fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 21:51:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aacd31cce6e48abbcbef22e4817fbe2\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06b789d4e9694a4090bea494bb63a0b9 : SUCCESS in 1m 54s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/28bce836c0c541e48b6c0a302e3986e2 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79eb461f785240c58dcfcebd5ddc6b95 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81d78dc0e5fd40bdb00f7b1f8aeb4b28 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04ab8e9ca7a74153b80b06527dc9abb1 : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c93b14342771cdf9a67ca7bfea6a285319a520f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:52:00.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"55c576939517489c404888715ccb0cebbba47d74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:53:43.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0f75ac6d830e666f8921d8472bff6efd9dc8b88c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 21:53:59.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"1ef2326befa65948fbf8df0ad2c553e3f3f21834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 22:12:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/029b2807f03e4351b11f4d2fdfac6573\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/935ab6057b47426da6cb577d0c0ff629 : SUCCESS in 3m 45s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c474f1dbc09e4495b93193a1cd07dc55 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1af3daa5eb4962bd0c616998f3ee18 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2684f074b2dc4cf0a53082825bf2a17c : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5fb62d5a8c841ea88cf70c055749c46 : SUCCESS in 6m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee50013058a468eb0b0ea4ca9ac8aee : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"42234b2bf43bf2778755dfa1ad10d346421f18a1","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-24 08:08:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7f84bbdacd5282a0ffb25475f2e3cd4e30318f4e","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-05-24 08:59:31.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6f1b693b169b5835103ae845d935940571f7e309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 09:39:31.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1678df58e4f4cc0faa5303d96e43081ece311791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 09:39:46.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f754679a98bd0d6e56a98014073d665b7414199e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 09:39:49.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4be71bdc7b17704eedb6813d3a526da0c4abcb8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-24 09:43:04.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"19eba02df1109680027e1db93c999daac2e3b7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 09:58:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0510868cde41a384fa90d8c8dbe18a\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/785a464b21df43de8381036a412004e0 : SUCCESS in 3m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a270cbf92b9d438095bfe8ca8fb83b16 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89b7bcae3d946a9a7d71f1e2d57215f : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e66162c0f38a402a80b379b632b74f65 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dfc5a17280c4f0095bb8a8b69e71fec : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"edc6e5d25f6b2a9286b1666b4464edb5f977f373","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-26 13:05:34.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"4e0cc681eefcfef8eafa125c9944f903fdc64d6d","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-07-04 15:15:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"590e74ed649bf1aee7d33e0764b52fb7f618e85f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 15:39:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"fa5515c43b8cdc9ce8a5f3d6d040b031384ed736","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 15:39:16.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"57fd632038a75ebcd6e6aed8f1ae49207a2f5ee7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 15:40:17.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"529f6fc33633fa226f0481385f9cdb0a0a2b25ce","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 15:40:52.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"51592a7c112618ab7dbdcbf3fd79c93621972e09","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 15:41:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0f7882b40a27cd79efe214b9c1439ef33178775c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 15:51:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf14ad505c344c4099fd1439bcb8396a\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7046f7e671e427f9451da0f9ebab32d : SUCCESS in 4m 37s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2721739f5c0741c89f84acc841b3706c : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166c15e731e94e9289b5f2ad1516ffd9 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a53f68944ae4e0f9ea513053e9723c4 : SUCCESS in 1m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/396da32f11dc409db07882358c63b2b6 : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4f13646a7eec9544a49cc4d4c9cfd41f6532255a","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-07-04 15:56:33.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a7adfd0789769699452ba30fc366c6a22385e92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 15:56:46.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9eb9a5f535ae6a98464da817443a20973e27a7ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:05:08.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe806962d2ab4ad4b5d1b53baeedc4bb\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3674c5fcfea445f921da9a39661d3e0 : SUCCESS in 4m 42s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/f495dbddf63f4be4ad94a46e6fbeef54 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1fba9cab474c1f99642d2452ee20ce : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d726e21106a34b5ab29490872b167864 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":9},{"id":"260782d7cc04b33ca826c3c0fcd826bbe9a060c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:05:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c0cfb260942102cfa341ccd3874c49af4e4cf145","revisions":{"7dd026cbec9962540d9643c73fff44999fdec247":{"kind":"REWORK","_number":1,"created":"2025-05-23 21:40:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/1"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/7dd026cbec9962540d9643c73fff44999fdec247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/7dd026cbec9962540d9643c73fff44999fdec247"}]},"branch":"refs/heads/master"},"2be5a2921d763a31a75c0ff488ac238caceef5fc":{"kind":"REWORK","_number":2,"created":"2025-05-23 21:45:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/2"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:45:23.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/2be5a2921d763a31a75c0ff488ac238caceef5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/2be5a2921d763a31a75c0ff488ac238caceef5fc"}]},"branch":"refs/heads/master"},"b1b59c372b0b43d3691f7ab1b244b353e923dc91":{"kind":"REWORK","_number":3,"created":"2025-05-23 21:52:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/3"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:52:00.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b1b59c372b0b43d3691f7ab1b244b353e923dc91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/b1b59c372b0b43d3691f7ab1b244b353e923dc91"}]},"branch":"refs/heads/master"},"ec6aca6aae4abcee31f4fefbd43a34b719fbba2e":{"kind":"REWORK","_number":4,"created":"2025-05-23 21:53:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/4"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:53:42.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ec6aca6aae4abcee31f4fefbd43a34b719fbba2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/ec6aca6aae4abcee31f4fefbd43a34b719fbba2e"}]},"branch":"refs/heads/master"},"f8917f66794102cf921402c654f0dca419897a9a":{"kind":"REWORK","_number":5,"created":"2025-05-24 09:39:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/5"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-24 09:38:20.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/f8917f66794102cf921402c654f0dca419897a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/f8917f66794102cf921402c654f0dca419897a9a"}]},"branch":"refs/heads/master"},"aed02794fbbec198a6606e9f255f42bb637bf869":{"kind":"REWORK","_number":6,"created":"2025-05-24 09:39:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/6"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-24 09:39:41.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/aed02794fbbec198a6606e9f255f42bb637bf869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/aed02794fbbec198a6606e9f255f42bb637bf869"}]},"branch":"refs/heads/master"},"e6846497a941f682506821bfe22a90449d5fd36b":{"kind":"REWORK","_number":7,"created":"2025-05-24 09:43:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/7"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-24 09:43:04.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/e6846497a941f682506821bfe22a90449d5fd36b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/e6846497a941f682506821bfe22a90449d5fd36b"}]},"branch":"refs/heads/master"},"46202ba748f20d429796c60683af02eb5e2df559":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-07-04 15:39:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/8"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 15:39:16.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46202ba748f20d429796c60683af02eb5e2df559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/46202ba748f20d429796c60683af02eb5e2df559"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0cfb260942102cfa341ccd3874c49af4e4cf145":{"kind":"REWORK","_number":9,"created":"2025-07-04 15:40:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/950802/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/02/950802/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/02/950802/9"}}},"commit":{"parents":[{"commit":"839bc57dc65553263c00d6e8818dbb01b5fafb77","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/839bc57dc65553263c00d6e8818dbb01b5fafb77"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-23 21:40:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 15:40:17.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I9d6a0d63f6948130e43fc0a6d3b5c89e3bbadf45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/c0cfb260942102cfa341ccd3874c49af4e4cf145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/c0cfb260942102cfa341ccd3874c49af4e4cf145"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
